)]}'
{"id":"openstack%2Fpubliccloud-sig~719182","triplet_id":"openstack%2Fpubliccloud-sig~master~I05c12768dd1648c59156bc8700238aaa59ee1dc3","project":"openstack/publiccloud-sig","branch":"master","topic":"cleanup-py27-support","hashtags":[],"change_id":"I05c12768dd1648c59156bc8700238aaa59ee1dc3","subject":"Cleanup py27 support and docs","status":"MERGED","created":"2020-04-11 14:21:24.000000000","updated":"2020-04-16 19:09:45.000000000","submitted":"2020-04-16 19:07:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719182-1587064070745-a68383b8","meta_rev_id":"a18fb9cde38763a3086ac7a330f1db84ae19d390","_number":719182,"virtual_id_number":719182,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23924,"name":"Tobias Rydberg","email":"tobias.rydberg@cleura.com","username":"tobberydberg"},{"value":0,"_account_id":10134,"name":"Sean Handley","email":"sean.handley@gmail.com","username":"seanhandley"},{"value":0,"_account_id":16984,"name":"Nick Jones","email":"nick@dischord.org","username":"yankcrime"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-16 19:07:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23924,"name":"Tobias Rydberg","email":"tobias.rydberg@cleura.com","username":"tobberydberg"},"all":[{"value":2,"date":"2020-04-16 18:59:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23924,"name":"Tobias Rydberg","email":"tobias.rydberg@cleura.com","username":"tobberydberg"},{"value":0,"_account_id":10134,"name":"Sean Handley","email":"sean.handley@gmail.com","username":"seanhandley"},{"value":0,"_account_id":16984,"name":"Nick Jones","email":"nick@dischord.org","username":"yankcrime"},{"value":1,"date":"2020-04-14 10:24:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23924,"name":"Tobias Rydberg","email":"tobias.rydberg@cleura.com","username":"tobberydberg"},"all":[{"value":1,"date":"2020-04-16 18:59:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23924,"name":"Tobias Rydberg","email":"tobias.rydberg@cleura.com","username":"tobberydberg"},{"value":0,"date":"2020-04-16 18:06:58.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10134,"name":"Sean Handley","email":"sean.handley@gmail.com","username":"seanhandley"},{"value":0,"date":"2020-04-16 18:06:58.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16984,"name":"Nick Jones","email":"nick@dischord.org","username":"yankcrime"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-16 18:06:58.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10134,"name":"Sean Handley","email":"sean.handley@gmail.com","username":"seanhandley"},{"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"},{"_account_id":16984,"name":"Nick Jones","email":"nick@dischord.org","username":"yankcrime"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23924,"name":"Tobias Rydberg","email":"tobias.rydberg@cleura.com","username":"tobberydberg"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-14 10:24:59.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-16 18:06:58.000000000","updated_by":{"_account_id":10134,"name":"Sean Handley","email":"sean.handley@gmail.com","username":"seanhandley"},"reviewer":{"_account_id":10134,"name":"Sean Handley","email":"sean.handley@gmail.com","username":"seanhandley"},"state":"REVIEWER"},{"updated":"2020-04-16 18:06:58.000000000","updated_by":{"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"},"reviewer":{"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"},"state":"REVIEWER"},{"updated":"2020-04-16 18:06:58.000000000","updated_by":{"_account_id":16984,"name":"Nick Jones","email":"nick@dischord.org","username":"yankcrime"},"reviewer":{"_account_id":16984,"name":"Nick Jones","email":"nick@dischord.org","username":"yankcrime"},"state":"REVIEWER"},{"updated":"2020-04-16 18:59:02.000000000","updated_by":{"_account_id":23924,"name":"Tobias Rydberg","email":"tobias.rydberg@cleura.com","username":"tobberydberg"},"reviewer":{"_account_id":23924,"name":"Tobias Rydberg","email":"tobias.rydberg@cleura.com","username":"tobberydberg"},"state":"REVIEWER"},{"updated":"2020-04-16 19:07:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f9843832b95e8d8754d59d39468ecb5a326c4d88","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-11 14:21:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00bb582eea6fc64c9f206efaf4771ce8f188821c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 14:25:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32078f4705c64c9d8b3fd048f3209b8c : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2c4d38a879cfdcfd74f385f7f8284855036aeecc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-14 10:24:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"157312816cd63d00900e34baff45b02b39592079","author":{"_account_id":23924,"name":"Tobias Rydberg","email":"tobias.rydberg@cleura.com","username":"tobberydberg"},"date":"2020-04-16 18:59:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"00b63cd068efa01916b3218fa6632cb012853fe6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 18:59:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0fc8d053b84d563fc1001a755c0829e3d461100b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 19:07:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3278fb0a9a36489b8383f58f473f5593 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"848796f677c50e8bca146ac0c7440780e1fdd9b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 19:07:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"32c45cf17d92d3f9de167272537ef790bc4d4533","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 19:09:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/3cb55d6c700e4175b12de484d6773e30 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a670331b47fd21078ed8dd4f26d526c2e0cbb99e","revisions":{"a670331b47fd21078ed8dd4f26d526c2e0cbb99e":{"kind":"REWORK","_number":1,"created":"2020-04-11 14:21:24.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/82/719182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/publiccloud-sig","ref":"refs/changes/82/719182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/publiccloud-sig refs/changes/82/719182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/publiccloud-sig refs/changes/82/719182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/publiccloud-sig refs/changes/82/719182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/publiccloud-sig refs/changes/82/719182/1"}}},"commit":{"parents":[{"commit":"ac9998f11f97df530fb738b196ea01f9aa261a1d","subject":"Merge \"Fix Review URL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/publiccloud-sig/commit/ac9998f11f97df530fb738b196ea01f9aa261a1d"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-11 14:18:29.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-11 14:20:28.000000000","tz":120},"subject":"Cleanup py27 support and docs","message":"Cleanup py27 support and docs\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Remove obsolete sections from setup.cfg\n- Cleanup doc/source/conf.py to remove now obsolete content.\n- Use newer openstackdocstheme version\n- Remove install_command from tox.ini, the default is fine\n- Remove unused file test-requirements.txt\n- Remove unused dependencies from requirements.txt, this repo\n  is docs only\n\nChange-Id: I05c12768dd1648c59156bc8700238aaa59ee1dc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/publiccloud-sig/commit/a670331b47fd21078ed8dd4f26d526c2e0cbb99e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/publiccloud-sig/commit/a670331b47fd21078ed8dd4f26d526c2e0cbb99e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
