)]}'
{"id":"openstack%2Fpuppet-aodh~654145","triplet_id":"openstack%2Fpuppet-aodh~master~Ie91e7e34423b1bf68443b75de0cf8097fcc8ff6b","project":"openstack/puppet-aodh","branch":"master","topic":"fix-REAME.md","hashtags":[],"change_id":"Ie91e7e34423b1bf68443b75de0cf8097fcc8ff6b","subject":"Fix Repository","status":"MERGED","created":"2019-04-21 05:00:50.000000000","updated":"2019-04-21 05:37:18.000000000","submitted":"2019-04-21 05:37:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654145-1555825038315-61dee02a","meta_rev_id":"109fff1d7c1d9dddeec573110cbedf3ae9a30957","_number":654145,"virtual_id_number":654145,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-04-21 05:37:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-04-21 05:32:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-04-21 05:32:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-21 05:37:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"598cf1fab9642df99c1e280235ff947d03f5926c","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 05:00:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07920091062627d9d4aefc3235cdb4ae723c8484","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 05:05:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/45/654145/1/check/puppet-openstack-lint-ubuntu-bionic/0110e01/ : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7cb02d012acb784e12256d69e9ac31489a3195b6","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 05:32:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"94cee453790dca8e6b0bcaa590bc22a8c21ae4ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 05:32:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7046a2fc441027122161c8d5df8f352e195d262e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 05:37:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"56ace4b4a3e89a3f0825124efa80c8199ce3e617","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 05:37:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/45/654145/1/gate/puppet-openstack-lint-ubuntu-bionic/a6dfac8/ : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e689c27b37f410d66f16cb8f3e00de154ecf77d","revisions":{"0e689c27b37f410d66f16cb8f3e00de154ecf77d":{"kind":"REWORK","_number":1,"created":"2019-04-21 05:00:50.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/45/654145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/45/654145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/45/654145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/45/654145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/45/654145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/45/654145/1"}}},"commit":{"parents":[{"commit":"43d7e4a0c478257fe436a93c8c8cc035fde9bad7","subject":"Replace git.openstack.org with opendev.org/openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/43d7e4a0c478257fe436a93c8c8cc035fde9bad7"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 04:59:08.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 04:59:08.000000000","tz":480},"subject":"Fix Repository","message":"Fix Repository\n\nChange-Id: Ie91e7e34423b1bf68443b75de0cf8097fcc8ff6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0e689c27b37f410d66f16cb8f3e00de154ecf77d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0e689c27b37f410d66f16cb8f3e00de154ecf77d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
