)]}'
{"id":"openstack%2Fpuppet-aodh~654198","triplet_id":"openstack%2Fpuppet-aodh~master~I4edb8c6cae3250c85718bdf85e966eb096079da3","project":"openstack/puppet-aodh","branch":"master","topic":"move-to-opendev","hashtags":[],"change_id":"I4edb8c6cae3250c85718bdf85e966eb096079da3","subject":"Move to opendev","status":"MERGED","created":"2019-04-21 06:45:41.000000000","updated":"2019-04-22 16:32:01.000000000","submitted":"2019-04-22 16:32:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654198-1555950721722-61dee02a","meta_rev_id":"5af97e32e4c09772d776bcce12edcbb93eb39c28","_number":654198,"virtual_id_number":654198,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-04-22 16:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-04-22 15:25:31.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-04-22 13:23:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-04-22 15:48:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-04-22 15:25:31.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2019-04-22 15:48:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 13:23:34.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-04-22 15:25:31.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-04-22 15:48:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-04-22 16:32:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a233f716520d112fd562b414c0d7ff101aa1a87a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 06:45:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"330f8c048bcbae540d1990d630826520f28b5093","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 08:01:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/98/654198/1/check/puppet-openstack-lint-ubuntu-bionic/f7eabe9/ : SUCCESS in 6m 36s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/98/654198/1/check/puppet-openstack-syntax-4-ubuntu-bionic/90f0776/ : SUCCESS in 7m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/98/654198/1/check/puppet-openstack-syntax-5-ubuntu-bionic/f57e93b/ : SUCCESS in 4m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/98/654198/1/check/puppet-openstack-syntax-6-ubuntu-bionic/91f095a/ : SUCCESS in 4m 20s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/98/654198/1/check/puppet-openstack-unit-4.8-centos-7/410f933/ : SUCCESS in 18m 14s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/98/654198/1/check/puppet-openstack-unit-5.5-centos-7/a23ca70/ : SUCCESS in 16m 49s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/98/654198/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/a6e2de0/ : SUCCESS in 8m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/98/654198/1/check/puppet-openstack-unit-latest-ubuntu-bionic/cfff5ef/ : FAILURE in 6m 48s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/98/654198/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/7bae4ba/ : FAILURE in 42m 41s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/98/654198/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/4083a5a/ : SUCCESS in 58m 47s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/98/654198/1/check/puppet-openstack-beaker-centos-7/36cc568/ : SUCCESS in 15m 59s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/98/654198/1/check/puppet-openstack-beaker-ubuntu-bionic/60aa3de/ : FAILURE in 14m 01s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/98/654198/1/check/build-openstack-releasenotes/e9be4aa/html/ : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"351bf71ac70a4873599cb64087ca445ba0289d55","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-04-22 13:23:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"594237cc36a46ea73aa0df5499fec385c5771050","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-04-22 15:25:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c31b92c4c955243462a7669e54bd2e401edbc2f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 15:28:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"117c49f68f33b3e720782fd3abc30fd514052017","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-22 15:48:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a34e7324d6f0ed5565ad692b460b4af047b37b4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:32:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8c62aab8a9b5341d59d82d9ea436f91f8b7d2a14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:32:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/98/654198/1/gate/puppet-openstack-lint-ubuntu-bionic/cf7edbb/ : SUCCESS in 4m 30s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/98/654198/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/51b4ecb/ : SUCCESS in 4m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/98/654198/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/8439074/ : SUCCESS in 4m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/98/654198/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/61e07c6/ : SUCCESS in 4m 01s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/98/654198/1/gate/puppet-openstack-unit-4.8-centos-7/ac3be66/ : SUCCESS in 10m 39s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/98/654198/1/gate/puppet-openstack-unit-5.5-centos-7/18f1d71/ : SUCCESS in 10m 35s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/98/654198/1/gate/puppet-openstack-unit-6.0-ubuntu-bionic/30d7320/ : SUCCESS in 7m 38s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/98/654198/1/gate/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/734ca35/ : FAILURE in 34m 59s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/98/654198/1/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/e97b72f/ : SUCCESS in 57m 20s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/98/654198/1/gate/puppet-openstack-beaker-centos-7/cd58665/ : SUCCESS in 15m 00s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/98/654198/1/gate/puppet-openstack-beaker-ubuntu-bionic/f1c2027/ : FAILURE in 10m 53s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/98/654198/1/gate/build-openstack-releasenotes/b4be4ef/html/ : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78d5b8869948d58bc53a392a590242b2abeda084","revisions":{"78d5b8869948d58bc53a392a590242b2abeda084":{"kind":"REWORK","_number":1,"created":"2019-04-21 06:45:41.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/98/654198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/98/654198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/98/654198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/98/654198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/98/654198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/98/654198/1"}}},"commit":{"parents":[{"commit":"0e689c27b37f410d66f16cb8f3e00de154ecf77d","subject":"Fix Repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0e689c27b37f410d66f16cb8f3e00de154ecf77d"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:48.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:48.000000000","tz":480},"subject":"Move to opendev","message":"Move to opendev\n\nChange-Id: I4edb8c6cae3250c85718bdf85e966eb096079da3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/78d5b8869948d58bc53a392a590242b2abeda084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/78d5b8869948d58bc53a392a590242b2abeda084"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
