)]}'
{"id":"openstack%2Fpuppet-aodh~656101","triplet_id":"openstack%2Fpuppet-aodh~master~Iaed4d4795a7bfa5871209bcd3b4a6d793476ac31","project":"openstack/puppet-aodh","branch":"master","topic":"bug/1826692","hashtags":[],"change_id":"Iaed4d4795a7bfa5871209bcd3b4a6d793476ac31","subject":"Deprecate idle_timeout option","status":"MERGED","created":"2019-04-28 03:00:41.000000000","updated":"2019-04-29 06:30:53.000000000","submitted":"2019-04-29 06:30:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656101-1556519453759-b3e99d47","meta_rev_id":"a6637db9374e150060a6d18eb5f71ff621fa7da2","_number":656101,"virtual_id_number":656101,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-04-29 04:18:12.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-04-29 06:30:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-04-28 08:41:31.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-04-28 17:17:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-04-28 17:17:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-28 08:41:31.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-04-28 17:17:28.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-04-29 06:30:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e444378f2af2e43e3d572b5d33930db1721def0","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-28 03:00:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01b5f4fe1a6e7106e71f3c1fa95a54f8432d488f","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-28 03:25:03.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"87d54d9e4a2e22cae52cbd128edab36b12fc0227","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 03:43:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/01/656101/2/check/puppet-openstack-lint-ubuntu-bionic/e15a378/ : SUCCESS in 3m 45s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/01/656101/2/check/puppet-openstack-syntax-4-ubuntu-bionic/ec5f673/ : SUCCESS in 6m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/01/656101/2/check/puppet-openstack-syntax-5-ubuntu-bionic/f3550e5/ : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/01/656101/2/check/puppet-openstack-syntax-6-ubuntu-bionic/9726ede/ : SUCCESS in 3m 39s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/01/656101/2/check/puppet-openstack-unit-4.8-centos-7/026f706/ : FAILURE in 10m 55s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/01/656101/2/check/puppet-openstack-unit-5.5-centos-7/4ea3cb0/ : FAILURE in 10m 10s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/01/656101/2/check/puppet-openstack-unit-6.0-ubuntu-bionic/6456ce6/ : FAILURE in 7m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/01/656101/2/check/puppet-openstack-unit-latest-ubuntu-bionic/c2f30be/ : FAILURE in 9m 43s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/01/656101/2/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/feb0087/ : FAILURE in 6m 02s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/01/656101/2/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/a7c12ae/ : FAILURE in 9m 05s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/01/656101/2/check/puppet-openstack-beaker-centos-7/401edf8/ : FAILURE in 8m 28s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/01/656101/2/check/puppet-openstack-beaker-ubuntu-bionic/22787a7/ : FAILURE in 8m 13s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/01/656101/2/check/build-openstack-releasenotes/b664aa2/html/ : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":2},{"id":"ac3f801ba217c5eb3ee7206148a1f5b63041a390","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-28 03:48:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8673cfe0cbf7c5d3dbcee85db1d54c20ef5e7586","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 05:04:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/01/656101/3/check/puppet-openstack-lint-ubuntu-bionic/5270365/ : SUCCESS in 4m 22s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/01/656101/3/check/puppet-openstack-syntax-4-ubuntu-bionic/2c27da7/ : SUCCESS in 4m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/01/656101/3/check/puppet-openstack-syntax-5-ubuntu-bionic/f9d8dcb/ : SUCCESS in 4m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/01/656101/3/check/puppet-openstack-syntax-6-ubuntu-bionic/f0a7902/ : SUCCESS in 6m 58s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/01/656101/3/check/puppet-openstack-unit-4.8-centos-7/a026a98/ : SUCCESS in 10m 47s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/01/656101/3/check/puppet-openstack-unit-5.5-centos-7/96a4f0e/ : SUCCESS in 12m 18s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/01/656101/3/check/puppet-openstack-unit-6.0-ubuntu-bionic/1766d47/ : SUCCESS in 8m 35s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/01/656101/3/check/puppet-openstack-unit-latest-ubuntu-bionic/584dc2d/ : SUCCESS in 8m 37s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/01/656101/3/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/f91c441/ : FAILURE in 30m 50s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/01/656101/3/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/0bf9187/ : SUCCESS in 54m 40s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/01/656101/3/check/puppet-openstack-beaker-centos-7/79f4fb1/ : SUCCESS in 15m 10s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/01/656101/3/check/puppet-openstack-beaker-ubuntu-bionic/3a0c321/ : FAILURE in 7m 46s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/01/656101/3/check/build-openstack-releasenotes/3838275/html/ : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":3},{"id":"38e4756c8dc87fcbd0a343179abfa9d3a2dd4a87","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-04-28 08:41:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c19c814c2d13a215668784107e2474db14da070d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-04-28 17:17:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1b3d15a92f2a4de8afd707323b2f5c8e579b74e5","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-29 04:18:12.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a7b589ac03590bebfe28742282d13a18352edb68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 05:33:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/01/656101/3/check/puppet-openstack-lint-ubuntu-bionic/a8c351f/ : SUCCESS in 3m 33s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/01/656101/3/check/puppet-openstack-syntax-4-ubuntu-bionic/2716e5a/ : SUCCESS in 3m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/01/656101/3/check/puppet-openstack-syntax-5-ubuntu-bionic/5997a49/ : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/01/656101/3/check/puppet-openstack-syntax-6-ubuntu-bionic/89f908e/ : SUCCESS in 3m 50s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/01/656101/3/check/puppet-openstack-unit-4.8-centos-7/8a8cde6/ : SUCCESS in 8m 54s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/01/656101/3/check/puppet-openstack-unit-5.5-centos-7/983c7e5/ : SUCCESS in 9m 50s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/01/656101/3/check/puppet-openstack-unit-6.0-ubuntu-bionic/dc6bd20/ : SUCCESS in 7m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/01/656101/3/check/puppet-openstack-unit-latest-ubuntu-bionic/e6fa853/ : SUCCESS in 8m 08s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/01/656101/3/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/527c7f1/ : FAILURE in 34m 03s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/01/656101/3/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/40dcd3d/ : SUCCESS in 1h 03m 57s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/01/656101/3/check/puppet-openstack-beaker-centos-7/79d5c87/ : SUCCESS in 14m 24s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/01/656101/3/check/puppet-openstack-beaker-ubuntu-bionic/9d7c934/ : FAILURE in 7m 04s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/01/656101/3/check/build-openstack-releasenotes/51998d9/html/ : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":3},{"id":"9f893f6a2677f097076e522a9d95d16aed015ad6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 05:33:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2c7a04e09b4d87331775d07c681b3ff03963e80c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 06:30:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a716745bab3a1b1d31f1a9ee95b8aaa398a48157","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 06:30:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/01/656101/3/gate/puppet-openstack-lint-ubuntu-bionic/bb1a003/ : SUCCESS in 3m 48s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/01/656101/3/gate/puppet-openstack-syntax-4-ubuntu-bionic/4bb6ea3/ : SUCCESS in 4m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/01/656101/3/gate/puppet-openstack-syntax-5-ubuntu-bionic/0187b3c/ : SUCCESS in 4m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/01/656101/3/gate/puppet-openstack-syntax-6-ubuntu-bionic/60062b5/ : SUCCESS in 3m 57s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/01/656101/3/gate/puppet-openstack-unit-4.8-centos-7/40b7c79/ : SUCCESS in 11m 51s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/01/656101/3/gate/puppet-openstack-unit-5.5-centos-7/b94f268/ : SUCCESS in 10m 28s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/01/656101/3/gate/puppet-openstack-unit-6.0-ubuntu-bionic/989b015/ : SUCCESS in 7m 44s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/01/656101/3/gate/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/c4664b7/ : FAILURE in 27m 48s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/01/656101/3/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/f562090/ : SUCCESS in 56m 01s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/01/656101/3/gate/puppet-openstack-beaker-centos-7/e7bf906/ : SUCCESS in 14m 00s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/01/656101/3/gate/puppet-openstack-beaker-ubuntu-bionic/bad497f/ : FAILURE in 7m 14s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/01/656101/3/gate/build-openstack-releasenotes/55220eb/html/ : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9b10d779b14daecd8772de338e31d4b6bb3c715e","revisions":{"22f3dc97429418846021c1dacc9f469fc85a4996":{"kind":"REWORK","_number":1,"created":"2019-04-28 03:00:41.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/01/656101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/01/656101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/01/656101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/01/656101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/01/656101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/01/656101/1"}}},"commit":{"parents":[{"commit":"32c07aaf0630f19d818805ba32f8ec025ce88714","subject":"Merge \"Provide more useful fail message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/32c07aaf0630f19d818805ba32f8ec025ce88714"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 02:58:49.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 02:58:49.000000000","tz":480},"subject":"Deprecate idle_timeout option","message":"Deprecate idle_timeout option\n\nThe idle_timeout parameter is deprecated, use connection_recycle_time\ninstead[1].\n\n[1]https://review.opendev.org/#/c/334182/\n\nChange-Id: Iaed4d4795a7bfa5871209bcd3b4a6d793476ac31\nCloses-Bug: #1826692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/22f3dc97429418846021c1dacc9f469fc85a4996"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/22f3dc97429418846021c1dacc9f469fc85a4996"}]},"branch":"refs/heads/master"},"3bfbf78d2560b858474ab24df938405c5458cd02":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-04-28 03:25:03.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/01/656101/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/01/656101/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/01/656101/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/01/656101/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/01/656101/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/01/656101/2"}}},"commit":{"parents":[{"commit":"32c07aaf0630f19d818805ba32f8ec025ce88714","subject":"Merge \"Provide more useful fail message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/32c07aaf0630f19d818805ba32f8ec025ce88714"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 02:58:49.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 03:20:22.000000000","tz":480},"subject":"Deprecate idle_timeout option","message":"Deprecate idle_timeout option\n\nThe idle_timeout parameter is deprecated, use connection_recycle_time\ninstead[1].\n\n[1]https://review.opendev.org/#/c/334182/\n\nChange-Id: Iaed4d4795a7bfa5871209bcd3b4a6d793476ac31\nDepends-On: https://review.opendev.org/656106/\nCloses-Bug: #1826692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3bfbf78d2560b858474ab24df938405c5458cd02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3bfbf78d2560b858474ab24df938405c5458cd02"}]},"branch":"refs/heads/master"},"9b10d779b14daecd8772de338e31d4b6bb3c715e":{"kind":"REWORK","_number":3,"created":"2019-04-28 03:48:14.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/01/656101/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/01/656101/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/01/656101/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/01/656101/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/01/656101/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/01/656101/3"}}},"commit":{"parents":[{"commit":"32c07aaf0630f19d818805ba32f8ec025ce88714","subject":"Merge \"Provide more useful fail message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/32c07aaf0630f19d818805ba32f8ec025ce88714"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 02:58:49.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 03:47:38.000000000","tz":480},"subject":"Deprecate idle_timeout option","message":"Deprecate idle_timeout option\n\nThe idle_timeout parameter is deprecated, use connection_recycle_time\ninstead[1].\n\n[1]https://review.opendev.org/#/c/334182/\n\nChange-Id: Iaed4d4795a7bfa5871209bcd3b4a6d793476ac31\nDepends-On: https://review.opendev.org/656106/\nCloses-Bug: #1826692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/9b10d779b14daecd8772de338e31d4b6bb3c715e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/9b10d779b14daecd8772de338e31d4b6bb3c715e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
