)]}'
{"id":"openstack%2Fpuppet-aodh~659410","triplet_id":"openstack%2Fpuppet-aodh~master~Id8ec77c7bbe59327d87f5bf583916e9123db1bc0","project":"openstack/puppet-aodh","branch":"master","topic":"add_openstackclient","hashtags":[],"change_id":"Id8ec77c7bbe59327d87f5bf583916e9123db1bc0","subject":"Add openstackclient installation to the client class","status":"MERGED","created":"2019-05-16 01:53:11.000000000","updated":"2019-05-17 20:25:26.000000000","submitted":"2019-05-17 20:25:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659410-1558124726984-172dd170","meta_rev_id":"691ed0528473f824b52d82b0ed81b19e39bdfcc6","_number":659410,"virtual_id_number":659410,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-17 20:25:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2019-05-17 09:54:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-05-17 19:21:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2019-05-17 19:21:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 09:54:59.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-05-17 19:21:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-17 20:25:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"693f975b1968dff5454d99be255f36c6d23cb1d8","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-16 01:53:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8940bd6dc4a2f6f9f1fe726d93efc0e650d9155b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 03:00:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/10/659410/1/check/puppet-openstack-lint-ubuntu-bionic/860fc67/ : SUCCESS in 4m 10s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/10/659410/1/check/puppet-openstack-syntax-4-ubuntu-bionic/787a520/ : SUCCESS in 3m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/10/659410/1/check/puppet-openstack-syntax-5-ubuntu-bionic/9f96128/ : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/10/659410/1/check/puppet-openstack-syntax-6-ubuntu-bionic/e844ff1/ : SUCCESS in 3m 35s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/10/659410/1/check/puppet-openstack-unit-4.8-centos-7/845cfb9/ : SUCCESS in 11m 48s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/10/659410/1/check/puppet-openstack-unit-5.5-centos-7/ebead9a/ : SUCCESS in 11m 08s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/10/659410/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/4bc90d6/ : SUCCESS in 6m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/10/659410/1/check/puppet-openstack-unit-latest-ubuntu-bionic/06658f8/ : SUCCESS in 7m 45s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/10/659410/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/4fd275b/ : FAILURE in 29m 41s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/10/659410/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/14e053b/ : SUCCESS in 1h 02m 30s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/10/659410/1/check/puppet-openstack-beaker-centos-7/5d34734/ : SUCCESS in 15m 46s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/10/659410/1/check/puppet-openstack-beaker-ubuntu-bionic/6fb1bc3/ : FAILURE in 8m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e96a4f0c3a6c1cf40e79d4a44dad9fd82d90624","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-05-16 10:26:48.000000000","message":"Patch Set 1: Code-Review-1\n\nMake sense, not sure why it\u0027s needed though. In the future we might deprecate the client classes once all python-*client are replaced by openstacksdk (iiuc).\n\nCan we get release note for all these changes? Since this is behavior change.","accounts_in_message":[],"_revision_number":1},{"id":"cb4a74f6d1ac228c2a1267f26c5f7c5164001a2e","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-17 01:01:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"403a0510ce09ec1f1ab84a2eddd2f8e80e4eae4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 02:03:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/10/659410/2/check/puppet-openstack-lint-ubuntu-bionic/adcb3a6/ : SUCCESS in 6m 13s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/10/659410/2/check/puppet-openstack-syntax-4-ubuntu-bionic/959e3fc/ : SUCCESS in 6m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/10/659410/2/check/puppet-openstack-syntax-5-ubuntu-bionic/b87e086/ : SUCCESS in 6m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/10/659410/2/check/puppet-openstack-syntax-6-ubuntu-bionic/0d6c9b8/ : SUCCESS in 6m 26s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/10/659410/2/check/puppet-openstack-unit-4.8-centos-7/f1e1ecb/ : SUCCESS in 17m 05s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/10/659410/2/check/puppet-openstack-unit-5.5-centos-7/99d233e/ : SUCCESS in 16m 41s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/10/659410/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/b661979/ : SUCCESS in 9m 59s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/10/659410/2/check/puppet-openstack-unit-latest-ubuntu-bionic/03fd6ee/ : SUCCESS in 9m 56s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/10/659410/2/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/2dfe38d/ : FAILURE in 28m 01s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/10/659410/2/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/606ac4b/ : SUCCESS in 55m 27s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/10/659410/2/check/puppet-openstack-beaker-centos-7/cff2d14/ : SUCCESS in 15m 28s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/10/659410/2/check/puppet-openstack-beaker-ubuntu-bionic/8bc497c/ : FAILURE in 9m 27s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/10/659410/2/check/build-openstack-releasenotes/00835e5/html/ : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":2},{"id":"93890fe3c7aaff36205049f992054ec131575ce1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-05-17 08:04:52.000000000","message":"Patch Set 2: Code-Review-1\n\nShould add a simple test as well in spec/classes/aodh_client_spec.rb\n\nit { is_expected.to contain_class(\u0027openstacklib::openstackclient\u0027) }","accounts_in_message":[],"_revision_number":2},{"id":"765e242d68a9090033e2d2f292c64ed7ccededcd","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-17 08:32:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bbcb03dab3d60ddf064159cbd6a638ba2428628c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 09:47:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/10/659410/3/check/puppet-openstack-lint-ubuntu-bionic/b31392e/ : SUCCESS in 3m 32s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/10/659410/3/check/puppet-openstack-syntax-4-ubuntu-bionic/5b9c1d2/ : SUCCESS in 3m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/10/659410/3/check/puppet-openstack-syntax-5-ubuntu-bionic/5a3f197/ : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/10/659410/3/check/puppet-openstack-syntax-6-ubuntu-bionic/7f402db/ : SUCCESS in 4m 40s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/10/659410/3/check/puppet-openstack-unit-4.8-centos-7/66934c0/ : SUCCESS in 32m 12s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/10/659410/3/check/puppet-openstack-unit-5.5-centos-7/791ddf7/ : SUCCESS in 10m 41s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/10/659410/3/check/puppet-openstack-unit-6.4-ubuntu-bionic/091cf2a/ : SUCCESS in 9m 48s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/10/659410/3/check/puppet-openstack-unit-latest-ubuntu-bionic/90cd9e4/ : SUCCESS in 6m 31s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/10/659410/3/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/a5dd937/ : FAILURE in 32m 43s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/10/659410/3/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/5496537/ : SUCCESS in 1h 06m 06s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/10/659410/3/check/puppet-openstack-beaker-centos-7/c0d6b5d/ : SUCCESS in 40m 39s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/10/659410/3/check/puppet-openstack-beaker-ubuntu-bionic/a96a739/ : FAILURE in 8m 00s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/10/659410/3/check/build-openstack-releasenotes/42bc321/html/ : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":3},{"id":"41a7d3fcca58622df5088ca2f53071034d52e123","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-05-17 09:54:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"62e71ddc74909201a8806f9ab156778eeb92016e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-17 19:21:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nSeems unnecessary because that should be handled via package dependencies","accounts_in_message":[],"_revision_number":3},{"id":"27aadcba9cdb0f6de064356f29e1c9d3fb299f2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 19:21:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7e4489949b2876d9629fd52a108dcc0cec79fc8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 20:25:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e21b73fc10e2dc6fc9e461e51f149f387c1166b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 20:25:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/10/659410/3/gate/puppet-openstack-lint-ubuntu-bionic/2d48070/ : SUCCESS in 4m 32s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/10/659410/3/gate/puppet-openstack-syntax-4-ubuntu-bionic/6765c25/ : SUCCESS in 5m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/10/659410/3/gate/puppet-openstack-syntax-5-ubuntu-bionic/42fde87/ : SUCCESS in 6m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/10/659410/3/gate/puppet-openstack-syntax-6-ubuntu-bionic/b02fbcb/ : SUCCESS in 3m 37s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/10/659410/3/gate/puppet-openstack-unit-4.8-centos-7/f0498f0/ : SUCCESS in 17m 10s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/10/659410/3/gate/puppet-openstack-unit-5.5-centos-7/9292427/ : SUCCESS in 17m 13s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/10/659410/3/gate/puppet-openstack-unit-6.4-ubuntu-bionic/56bfc38/ : SUCCESS in 14m 29s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/10/659410/3/gate/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/9a7dfe7/ : FAILURE in 33m 52s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/10/659410/3/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/f1583f2/ : SUCCESS in 54m 46s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/10/659410/3/gate/puppet-openstack-beaker-centos-7/2df7d4f/ : SUCCESS in 15m 02s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/10/659410/3/gate/puppet-openstack-beaker-ubuntu-bionic/f1a2c6c/ : FAILURE in 7m 36s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/10/659410/3/gate/build-openstack-releasenotes/d00785f/html/ : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d19b1209dc7d9cb28fc6e9302a0f9755e89488ce","revisions":{"c895fc282d01b4fb657e4e90cbdc947b7027cb0a":{"kind":"REWORK","_number":1,"created":"2019-05-16 01:53:11.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/10/659410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/10/659410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/10/659410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/10/659410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/10/659410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/10/659410/1"}}},"commit":{"parents":[{"commit":"9b10d779b14daecd8772de338e31d4b6bb3c715e","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/9b10d779b14daecd8772de338e31d4b6bb3c715e"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:22.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:22.000000000","tz":480},"subject":"Add openstackclient","message":"Add openstackclient\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: Id8ec77c7bbe59327d87f5bf583916e9123db1bc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c895fc282d01b4fb657e4e90cbdc947b7027cb0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c895fc282d01b4fb657e4e90cbdc947b7027cb0a"}]},"branch":"refs/heads/master"},"524b3beb6015a8d6034b31d4072e5e814ebc902f":{"kind":"REWORK","_number":2,"created":"2019-05-17 01:01:41.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/10/659410/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/10/659410/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/10/659410/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/10/659410/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/10/659410/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/10/659410/2"}}},"commit":{"parents":[{"commit":"9b10d779b14daecd8772de338e31d4b6bb3c715e","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/9b10d779b14daecd8772de338e31d4b6bb3c715e"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:22.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-17 00:59:53.000000000","tz":480},"subject":"Add openstackclient installation to the client class","message":"Add openstackclient installation to the client class\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: Id8ec77c7bbe59327d87f5bf583916e9123db1bc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/524b3beb6015a8d6034b31d4072e5e814ebc902f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/524b3beb6015a8d6034b31d4072e5e814ebc902f"}]},"branch":"refs/heads/master"},"d19b1209dc7d9cb28fc6e9302a0f9755e89488ce":{"kind":"REWORK","_number":3,"created":"2019-05-17 08:32:39.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/10/659410/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/10/659410/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/10/659410/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/10/659410/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/10/659410/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/10/659410/3"}}},"commit":{"parents":[{"commit":"9b10d779b14daecd8772de338e31d4b6bb3c715e","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/9b10d779b14daecd8772de338e31d4b6bb3c715e"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:22.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-17 08:31:13.000000000","tz":480},"subject":"Add openstackclient installation to the client class","message":"Add openstackclient installation to the client class\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: Id8ec77c7bbe59327d87f5bf583916e9123db1bc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d19b1209dc7d9cb28fc6e9302a0f9755e89488ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d19b1209dc7d9cb28fc6e9302a0f9755e89488ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
