)]}'
{"id":"openstack%2Fpuppet-aodh~661197","triplet_id":"openstack%2Fpuppet-aodh~master~I565ee8edd7770f002b32f08172b0ebcb7bd24e05","project":"openstack/puppet-aodh","branch":"master","topic":"memcached-inet6-prefix","hashtags":[],"change_id":"I565ee8edd7770f002b32f08172b0ebcb7bd24e05","subject":"Prefix memcached_servers IPv6 address with `inet6:`","status":"ABANDONED","created":"2019-05-24 06:31:59.000000000","updated":"2019-05-24 14:00:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7701ea7368e4af264ed28ea425323b34286f34fa","_number":661197,"virtual_id_number":661197,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-05-24 07:36:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 07:36:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eec3c02b7d173c359eee67b190a8d1d4347bbd64","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-05-24 06:31:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ea19130eaebac3d35ff472d4f6eb1451db0a8db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 07:36:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/97/661197/1/check/puppet-openstack-lint-ubuntu-bionic/0ce4492/ : SUCCESS in 6m 00s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/97/661197/1/check/puppet-openstack-syntax-4-ubuntu-bionic/86169f2/ : SUCCESS in 4m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/97/661197/1/check/puppet-openstack-syntax-5-ubuntu-bionic/596a12f/ : SUCCESS in 4m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/97/661197/1/check/puppet-openstack-syntax-6-ubuntu-bionic/2bd2130/ : SUCCESS in 5m 47s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/97/661197/1/check/puppet-openstack-unit-4.8-centos-7/b719d40/ : SUCCESS in 22m 57s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/97/661197/1/check/puppet-openstack-unit-5.5-centos-7/3a17ee3/ : SUCCESS in 14m 18s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/97/661197/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/2c8a882/ : SUCCESS in 14m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/97/661197/1/check/puppet-openstack-unit-latest-ubuntu-bionic/39da477/ : SUCCESS in 9m 57s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/97/661197/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/9e52611/ : FAILURE in 31m 25s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/97/661197/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/ca27991/ : SUCCESS in 58m 46s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/97/661197/1/check/puppet-openstack-beaker-centos-7/bb5444c/ : SUCCESS in 14m 04s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/97/661197/1/check/puppet-openstack-beaker-ubuntu-bionic/317bf45/ : FAILURE in 6m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25433e139e68a00a22539884ae1c43f111ebdfcc","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-05-24 14:00:42.000000000","message":"Abandoned\n\nWill patch keystone module only.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ed33c298e02e9387b0b7b64d6b3377a163a3100","revisions":{"3ed33c298e02e9387b0b7b64d6b3377a163a3100":{"kind":"REWORK","_number":1,"created":"2019-05-24 06:31:59.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/97/661197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/97/661197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/97/661197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/97/661197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/97/661197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/97/661197/1"}}},"commit":{"parents":[{"commit":"d19b1209dc7d9cb28fc6e9302a0f9755e89488ce","subject":"Add openstackclient installation to the client class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d19b1209dc7d9cb28fc6e9302a0f9755e89488ce"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-05-24 06:28:51.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-05-24 06:28:51.000000000","tz":120},"subject":"Prefix memcached_servers IPv6 address with `inet6:`","message":"Prefix memcached_servers IPv6 address with `inet6:`\n\nDepends-On: https://review.opendev.org/661136\nChange-Id: I565ee8edd7770f002b32f08172b0ebcb7bd24e05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3ed33c298e02e9387b0b7b64d6b3377a163a3100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3ed33c298e02e9387b0b7b64d6b3377a163a3100"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
