)]}'
{"id":"openstack%2Fpuppet-aodh~663200","triplet_id":"openstack%2Fpuppet-aodh~master~Icd43506338c56ea8bc0e4238ac3f520f9dc2a63a","project":"openstack/puppet-aodh","branch":"master","topic":"prep-release-train-1","hashtags":[],"change_id":"Icd43506338c56ea8bc0e4238ac3f520f9dc2a63a","subject":"Prepare Train M2","status":"MERGED","created":"2019-06-05 01:42:56.000000000","updated":"2019-06-10 13:21:04.000000000","submitted":"2019-06-10 13:21:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663200-1560172864265-2d20edbd","meta_rev_id":"be861aadd1b4fddc6d768996d59e51a751b52600","_number":663200,"virtual_id_number":663200,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-10 13:21:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-06-10 13:15:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-06-10 13:15:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 13:15:52.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-10 13:21:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"342304f03707e23b7f82490b4d831aa76c45b0da","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-06-05 01:42:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21193e2adb5c297a6d5a08cafe80e8a4a16fb067","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 02:05:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/00/663200/1/check/build-openstack-puppet-tarball/7123118/ : SUCCESS in 2m 59s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/00/663200/1/check/puppet-openstack-lint-ubuntu-bionic/5cb0e82/ : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f320e83a1c51f113717d451c8dc92ce56cd6d9f3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 19:59:54.000000000","message":"Patch Set 1:\n\nbtw this is the prep for m2 as the major version bump was m1","accounts_in_message":[],"_revision_number":1},{"id":"4a5dad5088c255ee0e45d30702dd09bba138c878","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-06-10 02:09:26.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2a803a1eb79dd7260cc42e5d4cad05aa5385c1e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 02:20:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/00/663200/2/check/build-openstack-puppet-tarball/aba1d66/ : SUCCESS in 2m 06s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/00/663200/2/check/puppet-openstack-lint-ubuntu-bionic/843197e/ : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":2},{"id":"e403f302a63286d03f4814bbf90e0fc36aeaac59","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 13:15:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"743d6440877e5bc0d4971cf84e890620d14dabc3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:16:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bf1604c22f96f2ca889e25d20e7522be29fb162e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:21:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5d1cfd0f73317cc0f87a3029cd6fc890f02e0c77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:21:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/00/663200/2/gate/build-openstack-puppet-tarball/3dbc5d0/ : SUCCESS in 2m 19s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/00/663200/2/gate/puppet-openstack-lint-ubuntu-bionic/c41c6d7/ : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"257be71d59f5487bb7208d19e5cded0c44357e6e","revisions":{"511f16a89cb997ed1d517c431814f90e512584aa":{"kind":"REWORK","_number":1,"created":"2019-06-05 01:42:56.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/00/663200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/00/663200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/00/663200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/00/663200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/00/663200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/00/663200/1"}}},"commit":{"parents":[{"commit":"d19b1209dc7d9cb28fc6e9302a0f9755e89488ce","subject":"Add openstackclient installation to the client class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d19b1209dc7d9cb28fc6e9302a0f9755e89488ce"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:15.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:15.000000000","tz":480},"subject":"Prepare Train M1","message":"Prepare Train M1\n\nUpdate the version metadata for Train milestone 1\n\nChange-Id: Icd43506338c56ea8bc0e4238ac3f520f9dc2a63a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/511f16a89cb997ed1d517c431814f90e512584aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/511f16a89cb997ed1d517c431814f90e512584aa"}]},"branch":"refs/heads/master"},"257be71d59f5487bb7208d19e5cded0c44357e6e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-10 02:09:26.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/00/663200/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/00/663200/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/00/663200/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/00/663200/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/00/663200/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/00/663200/2"}}},"commit":{"parents":[{"commit":"d19b1209dc7d9cb28fc6e9302a0f9755e89488ce","subject":"Add openstackclient installation to the client class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d19b1209dc7d9cb28fc6e9302a0f9755e89488ce"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:15.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-10 02:01:11.000000000","tz":480},"subject":"Prepare Train M2","message":"Prepare Train M2\n\nUpdate the version metadata for Train milestone 2\n\nChange-Id: Icd43506338c56ea8bc0e4238ac3f520f9dc2a63a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/257be71d59f5487bb7208d19e5cded0c44357e6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/257be71d59f5487bb7208d19e5cded0c44357e6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
