)]}'
{"id":"openstack%2Fpuppet-aodh~676556","triplet_id":"openstack%2Fpuppet-aodh~master~Idb50f1b4dd7e6975b725211420a4c1e87be8c542","project":"openstack/puppet-aodh","branch":"master","topic":"remove_deprecate_pki_related_options","hashtags":[],"change_id":"Idb50f1b4dd7e6975b725211420a4c1e87be8c542","subject":"Remove deprecated pki related options","status":"MERGED","created":"2019-08-15 03:52:34.000000000","updated":"2019-08-16 09:19:38.000000000","submitted":"2019-08-16 09:19:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676556-1565947178483-16970a7d","meta_rev_id":"23136801215954f66be3429e5baa6021a4469e5d","_number":676556,"virtual_id_number":676556,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-08-16 09:19:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-08-15 16:20:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-16 08:27:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-08-16 08:27:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 16:20:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-16 08:27:06.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-08-16 09:19:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f78d21642531f533fce42fe7256fed0d900718b","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-15 03:52:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71a2474dbe174bf2a433bf87bec9132bdfe9cfb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 04:50:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b71d9191af44555a9a8009f0f8f11c3 : SUCCESS in 5m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9f8cf2ee23947aba20ec965ffae9612 : SUCCESS in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bceefdbea2594eb08e2a899cfee6436f : SUCCESS in 4m 25s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/761fdfb06cc640b6ba80fb815d5f5d30 : SUCCESS in 14m 17s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b9d2d17ae874be295730bffba33d140 : SUCCESS in 9m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d988914b81b8431bab4b75f7df10d788 : SUCCESS in 12m 35s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/dc1e55b787f947e49e7d173edd8cb537 : FAILURE in 30m 48s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/0c73662951514514a21b02b7ea6be1a6 : SUCCESS in 50m 58s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d3916540f3e14fe7b04283510ef022ee : SUCCESS in 16m 27s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0497a31bf464062984e721e955392f0 : FAILURE in 7m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3328b12724904dd39dc364f5bd374d08 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f36f4db978ed3c85c658458dd0f154bcded7903e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-15 16:20:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"183a5f9cd42acbdfe268d7f96ad33286a975c37f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-16 08:27:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6985dd2bac12d66946be912c7f2eab416355ba8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 08:27:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26198ca5bb90f2ecbdfa1aca186744b87c9995ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 09:19:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"29bc4bf4032d0627f1c3c61db15cbc24f6a500f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 09:19:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21ce1746d1194b87a8e9fb7ceda6a2a9 : SUCCESS in 5m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/418a0dbdb49e4ffdb7de2561ef1081ee : SUCCESS in 4m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a2fa12fb6954093a90901e78d749eae : SUCCESS in 4m 28s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c2c6e9cfcc1b4d0882115678a5fea941 : SUCCESS in 13m 13s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e0944efc71442b28be9c783944bdf60 : SUCCESS in 12m 51s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9b23c337d0f041e1b199ede2952cfa38 : FAILURE in 33m 35s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/6492dca8aca44a99ace160eec3a3adfe : SUCCESS in 50m 57s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/13d1c63c163a4923af41f6caaaed9bdb : SUCCESS in 16m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43de82d3e9b14ca2a40a5ef73c59b75b : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e707d06dd5291c2fecb40d69c8c7e2a80bfa62a","revisions":{"0e707d06dd5291c2fecb40d69c8c7e2a80bfa62a":{"kind":"REWORK","_number":1,"created":"2019-08-15 03:52:34.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/56/676556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/56/676556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/56/676556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/56/676556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/56/676556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/56/676556/1"}}},"commit":{"parents":[{"commit":"dc67148434bd48e58196699ecb7c51d9b312ad20","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/dc67148434bd48e58196699ecb7c51d9b312ad20"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-15 03:51:37.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-15 03:51:37.000000000","tz":480},"subject":"Remove deprecated pki related options","message":"Remove deprecated pki related options\n\nThe deprecated pki related options check_revocations_for_cached and\nhash_algorithms option has been removed.\n\nChange-Id: Idb50f1b4dd7e6975b725211420a4c1e87be8c542\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0e707d06dd5291c2fecb40d69c8c7e2a80bfa62a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0e707d06dd5291c2fecb40d69c8c7e2a80bfa62a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
