)]}'
{"id":"openstack%2Fpuppet-aodh~677426","triplet_id":"openstack%2Fpuppet-aodh~master~I887d86893e90cebf5bc28dd539e42436f9a31c6a","project":"openstack/puppet-aodh","branch":"master","topic":"remove-deprecated","hashtags":[],"change_id":"I887d86893e90cebf5bc28dd539e42436f9a31c6a","subject":"Remove deprecated parameters","status":"MERGED","created":"2019-08-20 08:10:23.000000000","updated":"2019-08-21 01:27:03.000000000","submitted":"2019-08-21 01:27:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677426-1566350823275-ac68746e","meta_rev_id":"40a6df357e460692d3a3fed334f094376c28065d","_number":677426,"virtual_id_number":677426,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-21 01:27:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-08-21 00:28:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-08-20 15:39:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-08-21 00:28:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-20 15:39:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-21 00:28:40.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-08-21 01:27:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a034e13e34d9a70c6510cf1ab3b6ae188ddc429","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-20 08:10:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"613537a5d4c4b2b80abda3843de4bdb38bbb94ac","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-20 08:10:33.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"93105ec207cf1a0951d13a151f0d30fdf68135be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 09:11:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b806327249c4166980de0c0df38d16c : SUCCESS in 4m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1e075d033004b1dab3f8f3e9c0afdcc : SUCCESS in 7m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12358aae1ea942679f1549808cb1ca33 : SUCCESS in 4m 21s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4f6205f83b7c40698660962bf493acd2 : SUCCESS in 20m 18s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6875442a8f554a5eac82efe698c7ca82 : SUCCESS in 13m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/862d3f5918fe4856b37f6056188bce95 : SUCCESS in 7m 32s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/bfc13bca5fb7474483fa7d1034a39009 : FAILURE in 35m 22s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/5a31273db69744b5b65f1e062510ab8f : SUCCESS in 59m 16s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0818c80ce22a44ebbf02287eedd1d536 : SUCCESS in 17m 23s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bdef5c603a24d2d9c08d0a2aed83d76 : FAILURE in 7m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/727f7a37df4841148c85b0d5cde34e6f : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"f2fcc0f0a935d5eedb0e955342f945714005a48f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-20 15:39:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4380d4d8e06e0e06fd7b58bf636df9556aef08e8","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-21 00:28:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4c88937b3254f8510d99192402591de695027b3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 00:28:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f8f2aecb923906c7b24fbaf91f72a03b29fd2c08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 01:27:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"defd7cae3a13be9355eec1e1acaf50ac5ac2c06b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 01:27:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1da7f27b815e476e87a76c707c95c71f : SUCCESS in 6m 32s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e7af9b748954559b18b6aa0e07bc695 : SUCCESS in 5m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/006ec31adddc4f57858523182bed23b9 : SUCCESS in 5m 00s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7cefe5b6fbd04d04887f3550f6e927b9 : SUCCESS in 18m 29s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c34c3fc9437b4677bbde71dfb785dea9 : SUCCESS in 11m 51s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/34524df2cf084f76a0040ac2a03fa9cc : FAILURE in 31m 12s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/b1fe950173e9404f8a110cc47ee3dab2 : SUCCESS in 52m 23s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1d2d4bed0404405e9084182546be6038 : SUCCESS in 17m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b54e6c237a3f4cfcafe9fb39478bbcb1 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"06da1da22f049cd00c8d10270e821b33eeadf1ce","revisions":{"b5ddfe748be4c26fc68dc88d4cde267b86154e46":{"kind":"REWORK","_number":1,"created":"2019-08-20 08:10:23.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/26/677426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/26/677426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/26/677426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/26/677426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/26/677426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/26/677426/1"}}},"commit":{"parents":[{"commit":"0e707d06dd5291c2fecb40d69c8c7e2a80bfa62a","subject":"Remove deprecated pki related options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0e707d06dd5291c2fecb40d69c8c7e2a80bfa62a"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-08-20 08:10:06.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-08-20 08:10:06.000000000","tz":120},"subject":"Remove deprecated parameters","message":"Remove deprecated parameters\n\nRemove parameters that has been deprecated\nfor one cycle.\n\nChange-Id: I887d86893e90cebf5bc28dd539e42436f9a31c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b5ddfe748be4c26fc68dc88d4cde267b86154e46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b5ddfe748be4c26fc68dc88d4cde267b86154e46"}]},"branch":"refs/heads/master"},"06da1da22f049cd00c8d10270e821b33eeadf1ce":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-20 08:10:33.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/26/677426/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/26/677426/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/26/677426/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/26/677426/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/26/677426/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/26/677426/2"}}},"commit":{"parents":[{"commit":"0e707d06dd5291c2fecb40d69c8c7e2a80bfa62a","subject":"Remove deprecated pki related options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0e707d06dd5291c2fecb40d69c8c7e2a80bfa62a"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-08-20 08:10:06.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-08-20 08:10:24.000000000","tz":120},"subject":"Remove deprecated parameters","message":"Remove deprecated parameters\n\nRemove parameters that has been deprecated\nfor atleast one cycle.\n\nChange-Id: I887d86893e90cebf5bc28dd539e42436f9a31c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/06da1da22f049cd00c8d10270e821b33eeadf1ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/06da1da22f049cd00c8d10270e821b33eeadf1ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
