)]}'
{"id":"openstack%2Fpuppet-aodh~683742","triplet_id":"openstack%2Fpuppet-aodh~master~I881058fdc53c3ff381b96438ef46dfdc852ece94","project":"openstack/puppet-aodh","branch":"master","topic":"service_token_roles","hashtags":[],"change_id":"I881058fdc53c3ff381b96438ef46dfdc852ece94","subject":"Add support to configure service_token_roles in authtoken middleware","status":"MERGED","created":"2019-09-21 00:48:51.000000000","updated":"2019-09-23 08:48:04.000000000","submitted":"2019-09-23 08:46:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683742-1569228406587-c07f50fd","meta_rev_id":"519960b57b279b8909ba21acd800af2ae66cc37b","_number":683742,"virtual_id_number":683742,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-09-23 08:46:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-09-23 06:21:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-09-23 07:33:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-09-23 07:33:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 06:21:04.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-09-23 07:33:14.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-09-23 08:46:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0a17edb6542eb7f370b67380f93fc6b4109e98b1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-09-21 00:48:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3e1eab1d8ce809af73128697ccbbd241cbc2faf","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-09-21 01:06:18.000000000","message":"Patch Set 1: Code-Review-1\n\nit\u0027s better to add a release note :)","accounts_in_message":[],"_revision_number":1},{"id":"0d958805e48661b0e47342902e7bc5efc84983b7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-09-21 01:15:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"633fbad3a4ecb69c00488bc09c87fb3c0ca7e335","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-09-21 01:22:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"518a94873f92d4d303f28d33c5d69bfd255c8aac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 02:22:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78404e8dc1c64d9eb54e54216472de98 : SUCCESS in 4m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6438c68936f94c3183e3e41e17352454 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01f61eabae1b4be690335aa824b58b60 : SUCCESS in 4m 44s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/6523053661d7419cad3e22e2369ef191 : SUCCESS in 13m 25s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8762d083004a4de8a0f47f51812a0c0b : SUCCESS in 9m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4627613dad54428f869a239eafdfa808 : SUCCESS in 9m 00s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a5f3d033985e41119d1f16c5ea73e18b : FAILURE in 33m 52s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/9c9a710b1de54ca49153e3073c8c0a76 : SUCCESS in 53m 19s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b29c5ea652a342c5bdea5287074cacf5 : SUCCESS in 16m 15s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2cdba441f2d8426d96717b22c9906802 : FAILURE in 8m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bf8619a6d764b6bbde9a6413a015991 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":3},{"id":"ba86dd1e9573e9e00d966629b767bfee515c4b84","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-09-23 06:21:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d82a5c53d7f47145efaf2e5b14d5c5a5ddbc792e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-09-23 07:33:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3cd956636fa23f202f15c30a960edada6ec9648e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 07:33:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"114c3b16dccdd531eb9122ef3499b00efcfcd4b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 08:46:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a795a345392fb810a434af9986f137e6f29f806e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 08:46:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e67285b632b48629c6ff00ebd343192 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aecc2a8df99a45ec89c7b229311a6647 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c57c845306ed4986bc03288f146c3bd3 : SUCCESS in 4m 45s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3e0f548e440a4eb08582a6068b5d536f : SUCCESS in 17m 56s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d4d3a911bd14e468980d0688236f51f : SUCCESS in 11m 07s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/88c35800e2fd46aca9338ae024c81513 : FAILURE in 40m 49s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/f3250a86c9804b54aa660f4512a5370a : SUCCESS in 1h 11m 42s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/44aad59ccfe249aeb664425ca28dfc84 : SUCCESS in 17m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ed33caf340141cdaad3c469d39e2885 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"a9485e1e9da5ba95444523d2753e741d99277a0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 08:48:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fb90940fa4746c79760eb37a123f612 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f9491ff8dfdf55ea8051e39bd9203c1d1ae352aa","revisions":{"4e56898a5289f6b2a2a0569db982c4c8ae43c4f5":{"kind":"REWORK","_number":1,"created":"2019-09-21 00:48:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/683742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/42/683742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/683742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/683742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/683742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/42/683742/1"}}},"commit":{"parents":[{"commit":"2b0cfc8fa4f3bc08c7f190c7e922e9ab79ca836b","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/2b0cfc8fa4f3bc08c7f190c7e922e9ab79ca836b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-09-21 00:48:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-09-21 00:48:49.000000000","tz":540},"subject":"Add support to configure service_token_roles in authtoken middleware","message":"Add support to configure service_token_roles in authtoken middleware\n\nChange-Id: I881058fdc53c3ff381b96438ef46dfdc852ece94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/4e56898a5289f6b2a2a0569db982c4c8ae43c4f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/4e56898a5289f6b2a2a0569db982c4c8ae43c4f5"}]},"branch":"refs/heads/master"},"f8d123602980eccd3e7975b86590dedb15b24bd4":{"kind":"REWORK","_number":2,"created":"2019-09-21 01:15:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/683742/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/42/683742/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/683742/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/683742/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/683742/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/42/683742/2"}}},"commit":{"parents":[{"commit":"2b0cfc8fa4f3bc08c7f190c7e922e9ab79ca836b","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/2b0cfc8fa4f3bc08c7f190c7e922e9ab79ca836b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-09-21 00:48:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-09-21 01:15:29.000000000","tz":540},"subject":"Add support to configure service_token_roles in authtoken middleware","message":"Add support to configure service_token_roles in authtoken middleware\n\nChange-Id: I881058fdc53c3ff381b96438ef46dfdc852ece94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f8d123602980eccd3e7975b86590dedb15b24bd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f8d123602980eccd3e7975b86590dedb15b24bd4"}]},"branch":"refs/heads/master"},"f9491ff8dfdf55ea8051e39bd9203c1d1ae352aa":{"kind":"REWORK","_number":3,"created":"2019-09-21 01:22:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/683742/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/42/683742/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/683742/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/683742/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/683742/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/42/683742/3"}}},"commit":{"parents":[{"commit":"2b0cfc8fa4f3bc08c7f190c7e922e9ab79ca836b","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/2b0cfc8fa4f3bc08c7f190c7e922e9ab79ca836b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-09-21 00:48:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-09-21 01:22:19.000000000","tz":540},"subject":"Add support to configure service_token_roles in authtoken middleware","message":"Add support to configure service_token_roles in authtoken middleware\n\nChange-Id: I881058fdc53c3ff381b96438ef46dfdc852ece94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f9491ff8dfdf55ea8051e39bd9203c1d1ae352aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f9491ff8dfdf55ea8051e39bd9203c1d1ae352aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
