)]}'
{"id":"openstack%2Fpuppet-aodh~684465","triplet_id":"openstack%2Fpuppet-aodh~master~Ifdb5390e9a3cb989c05c6e7fd97b442fa7ac16c1","project":"openstack/puppet-aodh","branch":"master","topic":"prep-release-train-rc1","hashtags":[],"change_id":"Ifdb5390e9a3cb989c05c6e7fd97b442fa7ac16c1","subject":"Prepare Train RC1","status":"MERGED","created":"2019-09-25 05:46:52.000000000","updated":"2019-09-27 01:56:57.000000000","submitted":"2019-09-27 01:56:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684465-1569549417744-dadc74e4","meta_rev_id":"f987611a8dc3b59d84af9ff0bb2d2a4a8b9f2e22","_number":684465,"virtual_id_number":684465,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 01:56:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-27 01:46:36.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-27 01:46:36.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 01:46:36.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-27 01:56:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5dcbe6f2934927de3ec630dae106796b119c2f9","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-09-25 05:46:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5c69696610ec0b7d1c03a3949d9efeb8ad61d7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 05:56:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/ff9e584b97ef4067a0bb4efd7e2f4eee : SUCCESS in 2m 01s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4b8757ad3ac4cdb8052f9e9b5d81b48 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e6d6b8c870e8f3772fa323e09f7315fd11f322c0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-27 01:46:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4ca3ab067ac43ec1a4e2d249e546a606aea1c96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:46:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a48f8ea9facb3a5d41804ee2288a01eaeccf0deb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:56:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a8fc6611cb3a33f3c93333510e3c5607299e31dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:56:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/82ff2fb767794789a1b58999ebb2cd75 : SUCCESS in 3m 10s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf43baf7ccff4ea9a3476f4738e1483d : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62ffa024bcf97722fa3ccca53e8efa85835dfffe","revisions":{"62ffa024bcf97722fa3ccca53e8efa85835dfffe":{"kind":"REWORK","_number":1,"created":"2019-09-25 05:46:52.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/65/684465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/65/684465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/65/684465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/65/684465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/65/684465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/65/684465/1"}}},"commit":{"parents":[{"commit":"f9491ff8dfdf55ea8051e39bd9203c1d1ae352aa","subject":"Add support to configure service_token_roles in authtoken middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f9491ff8dfdf55ea8051e39bd9203c1d1ae352aa"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-09-25 05:45:42.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-09-25 05:45:42.000000000","tz":480},"subject":"Prepare Train RC1","message":"Prepare Train RC1\n\nUpdate the version metadata for Train milestone RC1\n\nChange-Id: Ifdb5390e9a3cb989c05c6e7fd97b442fa7ac16c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/62ffa024bcf97722fa3ccca53e8efa85835dfffe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/62ffa024bcf97722fa3ccca53e8efa85835dfffe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
