)]}'
{"id":"openstack%2Fpuppet-aodh~684926","triplet_id":"openstack%2Fpuppet-aodh~master~Ie12f71fdee11cadd623327a450858fc3b1c3ccab","project":"openstack/puppet-aodh","branch":"master","topic":"constraints","hashtags":[],"change_id":"Ie12f71fdee11cadd623327a450858fc3b1c3ccab","subject":"Update the constraints url","status":"MERGED","created":"2019-09-26 05:45:42.000000000","updated":"2019-09-27 23:31:11.000000000","submitted":"2019-09-27 23:30:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684926-1569627007040-654589fa","meta_rev_id":"5fa05fa889dabebaa263902a88dd526953a5f03a","_number":684926,"virtual_id_number":684926,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-09-27 21:28:55.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 23:30:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-27 01:46:39.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-27 01:46:39.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 01:46:39.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-27 21:28:55.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-27 23:30:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e7b87c0f45cdd810d51713870ce292724d67026c","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-26 05:45:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cd02960a93a185b73c3612d7e58e1f9480c897f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 07:03:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ccbe04bc5814c118d3c9d8930ae6fe5 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b91a27aa299415eb9ca199bc735541a : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c07092bbf9434bada6de32b2eee1e78b : SUCCESS in 5m 03s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/760dbb62d602420b998689a410e294d3 : SUCCESS in 11m 35s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7921b6989c414fe686579eb8a8015a9d : SUCCESS in 8m 34s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76794bfdf31e46b69713658294d57ed1 : SUCCESS in 9m 17s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/da3c94f91def4cc19eb437313892f692 : FAILURE in 32m 52s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/0ba0026b40db4d6e96c3b2f5a40fc8ba : SUCCESS in 54m 14s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/566b26f297c34368be65796bed715b3f : SUCCESS in 18m 09s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e85a16ddc6664dfd897c00154163fce6 : FAILURE in 7m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d360c18c1cd045a4874608ffa501b21a : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d75115a9175beb9340538397ffea0e56c396d8aa","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-27 01:46:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8772df164127d0ccf4718b4f212e50dfcbe2943e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:46:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7730f459435227d909b0d1e543417446df2dd91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 02:36:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca5cd483ca8147b69fcb5a7c19281bae : SUCCESS in 6m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9edbdb3753fa4f6f8d9bdb2c570b5bda : SUCCESS in 4m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac3ab98f5ba54958bfb9befae6fac066 : SUCCESS in 7m 17s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/6482470c0bab4abb8a4da90c91ccaa98 : SUCCESS in 17m 01s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24f8bb363df2438eb558514135428177 : SUCCESS in 15m 59s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d62bf4036626499590a1a5014ba96eb0 : FAILURE in 41m 40s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/df40b605814b4677896c463224706119 : FAILURE in 47m 38s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c4991d109daf48c086eadaf01214adc2 : SUCCESS in 23m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81d5f6b1fdb74912936cc949e3c3b040 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"4571efa694b3f13b1c6840b54a633ae6f648be79","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-27 21:28:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a52e5fa8890edc7474c97214ad89ec022ffa9a58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 22:20:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11fde1f4f49242ff803204fd17c39a0f : SUCCESS in 4m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7696809692c64a97a437ed4473c663c9 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55c291c667574763aa2b30bc38b44f50 : SUCCESS in 3m 52s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/933054f9773a4a049c65589ce6d209f5 : SUCCESS in 15m 47s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ada7021546b2422a9127a590ba9ab90a : SUCCESS in 10m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19517c49a0bb457f8e6569989e8765be : SUCCESS in 8m 33s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/aa3e44ff042e4a50a408e5828c927bd0 : FAILURE in 29m 39s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/16465eaf2c1348d8bf371026970e08e7 : SUCCESS in 47m 19s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5836794aada649bb8dbe169c618a1332 : SUCCESS in 16m 49s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/442b6a54d7d8487988e31358c16ff438 : FAILURE in 9m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96adf32778234b68bbbd3b9d7e575628 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8d660e35f2da75a17ac91c0262fe798282975ae4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 22:21:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad9fd78457b64dd595c8b75981d064c839a3d169","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 23:30:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da122b6552d44339a90c2dbb685d23c3 : SUCCESS in 5m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c34984c16f2467f92be615310e5d865 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b50b43dfca1e4964aaed4299822489bf : SUCCESS in 5m 52s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/06edd5ebac544557bde985a62c31a1fd : SUCCESS in 14m 36s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/477074f9ec774d17b1224204b7c2fb4a : SUCCESS in 13m 01s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1b450670275044198448003ce0bad580 : FAILURE in 47m 52s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/11ccbd82220c4592b0574d28a406a9c5 : SUCCESS in 1h 05m 46s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a659f011646a40398ccc042275bf924e : SUCCESS in 18m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac29fd6aa7fb4dab8369e5a2b893d853 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"b424d84a8dc3822a0a7390a28324410ea930c79a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 23:30:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"96e0a6fa85f0810633efd821ce9441975c14df51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 23:31:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec22a920b9bb4688b323256a6f046433 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d3b3f45e5b2c44f45709a5e3412013a28c25057","revisions":{"9d3b3f45e5b2c44f45709a5e3412013a28c25057":{"kind":"REWORK","_number":1,"created":"2019-09-26 05:45:42.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/26/684926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/26/684926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/26/684926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/26/684926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/26/684926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/26/684926/1"}}},"commit":{"parents":[{"commit":"f9491ff8dfdf55ea8051e39bd9203c1d1ae352aa","subject":"Add support to configure service_token_roles in authtoken middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f9491ff8dfdf55ea8051e39bd9203c1d1ae352aa"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 05:45:20.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 05:46:26.000000000","tz":480},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Ie12f71fdee11cadd623327a450858fc3b1c3ccab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/9d3b3f45e5b2c44f45709a5e3412013a28c25057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/9d3b3f45e5b2c44f45709a5e3412013a28c25057"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
