)]}'
{"id":"openstack%2Fpuppet-aodh~693342","triplet_id":"openstack%2Fpuppet-aodh~master~I8261397bfee8d7e48c2d857485a9504620c65c35","project":"openstack/puppet-aodh","branch":"master","topic":"notif","hashtags":[],"change_id":"I8261397bfee8d7e48c2d857485a9504620c65c35","subject":"WIP Switch aodh_config ini management to ruby backend","status":"ABANDONED","created":"2019-11-07 11:27:19.000000000","updated":"2021-08-18 12:03:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fb1e68cf2a6d1c83eaf8641c443e6eeed26409bb","_number":693342,"virtual_id_number":693342,"owner":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-22 17:34:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-04 08:34:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62017bf14556a6026a6ca1637e7ade5fdaf60f51","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-11-07 11:27:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46db7b82a0b529d2a473631826cff118b2ce62b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 12:19:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67e9d1f7ee7f4ceebd6f98ac3f609e9b : SUCCESS in 4m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de3d294baeb24caab6ac09bae89cf159 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cded58b7a4784b7db0f1813cb9d28242 : SUCCESS in 4m 10s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/9a991a2e46014259af5a5e81853db9dc : FAILURE in 13m 55s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9bf1fbed3b244a1b8df37d61643b59b5 : FAILURE in 9m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b327a04400b4b4fbbf214cb13a1cdf5 : FAILURE in 8m 49s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/272badc101e24b68aff380518fe787ff : SUCCESS in 43m 04s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/da5634918a8e420cb9bad6371d034ca3 : SUCCESS in 51m 17s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f480bedcfbe94b2f8c9ca803f28dd970 : SUCCESS in 16m 38s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e7b616e32a34d2a94296354bc4f5ac2 : FAILURE in 17m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1fb613cd40c2f744d2906a8f8848e594a3d7f70","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-05-01 14:46:25.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"687da87ab2f4ddfc39377eaaeb2be2c76d5ab9be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 15:50:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55ab640ea0e74dd0a65acceca411ec6e : SUCCESS in 4m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db97697067784edb92df339ea2c3c2d3 : SUCCESS in 5m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6a4ec7060f744559876d7408e7efae4 : SUCCESS in 5m 52s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/da8ddb15a47f4a20a2bd102f76d183f0 : FAILURE in 12m 49s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/fde98a0b08614574a8ab42a20457db53 : FAILURE in 12m 06s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/846c3767fcf74b6b8b233a644c9b7fd5 : FAILURE in 10m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36d15e1d68ef4a69a231e1b622a8cf6e : FAILURE in 10m 34s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/671c563d8155413f998f7b6278165f70 : SUCCESS in 54m 12s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/427a0335c641485ea140e818c5ac04a0 : SUCCESS in 55m 06s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e081a3def4fc442e9aa3c5b6d40eddec : SUCCESS in 19m 06s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/0856db610f5f49e9abdb6d42f1ecdc4c : FAILURE in 8m 55s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea9f28b465fa4643aaadae0b7819ac92 : FAILURE in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"83e89c48734f5eda3ba7e638cd73b20a1297f963","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-07-04 07:32:42.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c6d5ef982257bee51ddb527532a743e3112f52e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 08:34:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1a3233fce814bb88c7b0a67c2748bfe : SUCCESS in 6m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a51ad7c939b345d594d78a1b1d17660b : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/625bd2e14cf94e469da1888f4d389278 : SUCCESS in 5m 19s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3a61335d571f407eb862f64ddfe1ff6d : FAILURE in 16m 53s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/53b467b90e57408db43d654d9f0a66da : FAILURE in 17m 01s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09acc71366f549e0b192ad6e096c395e : FAILURE in 9m 02s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39d9a9dec4054797b1f34f82f400d611 : FAILURE in 11m 50s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e097e514e80d4159aacc966ddb68cb41 : SUCCESS in 56m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6f0e2dd827024c7c8bdf4cb2453c7198 : SUCCESS in 59m 50s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/167fc55cb5d947629391facd14bc948c : SUCCESS in 22m 25s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/99b56a5b7f594f58a396092b22bae2d4 : FAILURE in 10m 33s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0b4b98961364eb5ae1820a56d4780be : FAILURE in 6m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"11710d93e09a7b5bc1229b6685def4266cf1c849","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-11-22 16:26:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e087b314feaa71e6cc2a35927d3db34482eaea54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-22 17:34:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62da81f34f4e4d9ba94be00f1547d9b6 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbd5a84827e54021aa5df9db83974ddd : SUCCESS in 3m 16s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/15960596db014646bd6cff5ef6bf303c : FAILURE in 11m 21s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28cf780ee6f742c0890d400753ebea3f : FAILURE in 9m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41c1fff371bd4234a786be4aefcfeba0 : FAILURE in 9m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b69462a7acba4da793d3109cd65eab31 : SUCCESS in 49m 51s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/91c0d5a90c214d45a16857e425c038c8 : SUCCESS in 1h 03m 41s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d8aee3a40f18470da854619566ad06e1 : SUCCESS in 18m 35s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6dcb34546afe40a6a2c406c9ae5200b2 : SUCCESS in 12m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fb1e68cf2a6d1c83eaf8641c443e6eeed26409bb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-08-18 12:03:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"acd0e4cf4b0d65b24853190aa723b51913bf5f30","revisions":{"f80d4e762a0ee31a7454ab851f505a69a9ef04de":{"kind":"REWORK","_number":1,"created":"2019-11-07 11:27:19.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/42/693342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/42/693342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/693342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/693342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/693342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/42/693342/1"}}},"commit":{"parents":[{"commit":"a8fec741ead2f86e0220e2e79fb58a9ea11a37c6","subject":"Fix rabbitmq dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/a8fec741ead2f86e0220e2e79fb58a9ea11a37c6"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-11-07 11:26:39.000000000","tz":60},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-11-07 11:27:17.000000000","tz":60},"subject":"WIP Switch aodh_config ini management to ruby backend","message":"WIP Switch aodh_config ini management to ruby backend\n\nThis is done so that we can correctly specify multiple notification\ndrivers. Via puppet-oslo change in If65946412b42e0919456ed92fdd8e3788ad67872\nwe want to be able to specify multiple drivers for oslo messaging\nnotification by passing a list.\n\nLet\u0027s switch the ini management backend to the ruby backend so we can\ncorrectly have multiple driver\u003d lines.\n\nNB: WIP as I need to do another round of tests (initial tests were fine)\n\nChange-Id: I8261397bfee8d7e48c2d857485a9504620c65c35\nRelated-Bug: #1851629\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f80d4e762a0ee31a7454ab851f505a69a9ef04de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f80d4e762a0ee31a7454ab851f505a69a9ef04de"}]},"branch":"refs/heads/master"},"58fe1479adaa1e5f192ef19a1b127a9f92b67f45":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-01 14:46:25.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/42/693342/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/42/693342/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/693342/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/693342/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/693342/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/42/693342/2"}}},"commit":{"parents":[{"commit":"f34c57e5b142178afb4e986a84823980b37e9fcb","subject":"Bump openstackdocstheme to 1.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f34c57e5b142178afb4e986a84823980b37e9fcb"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-11-07 11:26:39.000000000","tz":60},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-05-01 14:46:15.000000000","tz":120},"subject":"WIP Switch aodh_config ini management to ruby backend","message":"WIP Switch aodh_config ini management to ruby backend\n\nThis is done so that we can correctly specify multiple notification\ndrivers. Via puppet-oslo change in If65946412b42e0919456ed92fdd8e3788ad67872\nwe want to be able to specify multiple drivers for oslo messaging\nnotification by passing a list.\n\nLet\u0027s switch the ini management backend to the ruby backend so we can\ncorrectly have multiple driver\u003d lines.\n\nNB: WIP as I need to do another round of tests (initial tests were fine)\n\nChange-Id: I8261397bfee8d7e48c2d857485a9504620c65c35\nRelated-Bug: #1851629\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/58fe1479adaa1e5f192ef19a1b127a9f92b67f45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/58fe1479adaa1e5f192ef19a1b127a9f92b67f45"}]},"branch":"refs/heads/master"},"7b7aa112517f4c7a8a42cea6da16af751f747554":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-04 07:32:42.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/42/693342/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/42/693342/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/693342/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/693342/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/693342/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/42/693342/3"}}},"commit":{"parents":[{"commit":"5329372ad3b71d94383ae3c95b022408566b71eb","subject":"Prepare Victoria M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/5329372ad3b71d94383ae3c95b022408566b71eb"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-11-07 11:26:39.000000000","tz":60},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-07-04 07:32:42.000000000","tz":0},"subject":"WIP Switch aodh_config ini management to ruby backend","message":"WIP Switch aodh_config ini management to ruby backend\n\nThis is done so that we can correctly specify multiple notification\ndrivers. Via puppet-oslo change in If65946412b42e0919456ed92fdd8e3788ad67872\nwe want to be able to specify multiple drivers for oslo messaging\nnotification by passing a list.\n\nLet\u0027s switch the ini management backend to the ruby backend so we can\ncorrectly have multiple driver\u003d lines.\n\nNB: WIP as I need to do another round of tests (initial tests were fine)\n\nChange-Id: I8261397bfee8d7e48c2d857485a9504620c65c35\nRelated-Bug: #1851629\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/7b7aa112517f4c7a8a42cea6da16af751f747554"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/7b7aa112517f4c7a8a42cea6da16af751f747554"}]},"branch":"refs/heads/master"},"acd0e4cf4b0d65b24853190aa723b51913bf5f30":{"kind":"REWORK","_number":4,"created":"2020-11-22 16:26:51.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/42/693342/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/42/693342/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/693342/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/693342/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/42/693342/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/42/693342/4"}}},"commit":{"parents":[{"commit":"5cfdef76a056b510a82fad7a98d2048fbb32f8be","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/5cfdef76a056b510a82fad7a98d2048fbb32f8be"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-11-07 11:26:39.000000000","tz":60},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-22 16:26:42.000000000","tz":60},"subject":"WIP Switch aodh_config ini management to ruby backend","message":"WIP Switch aodh_config ini management to ruby backend\n\nThis is done so that we can correctly specify multiple notification\ndrivers. Via puppet-oslo change in If65946412b42e0919456ed92fdd8e3788ad67872\nwe want to be able to specify multiple drivers for oslo messaging\nnotification by passing a list.\n\nLet\u0027s switch the ini management backend to the ruby backend so we can\ncorrectly have multiple driver\u003d lines.\n\nNB: WIP as I need to do another round of tests (initial tests were fine)\n\nChange-Id: I8261397bfee8d7e48c2d857485a9504620c65c35\nRelated-Bug: #1851629\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/acd0e4cf4b0d65b24853190aa723b51913bf5f30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/acd0e4cf4b0d65b24853190aa723b51913bf5f30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
