)]}'
{"id":"openstack%2Fpuppet-aodh~707528","triplet_id":"openstack%2Fpuppet-aodh~master~I2cc75c80e2b951bd6ca7b27b507461f8f358c4cb","project":"openstack/puppet-aodh","branch":"master","topic":"prep-release-ussuri-2","hashtags":[],"change_id":"I2cc75c80e2b951bd6ca7b27b507461f8f358c4cb","subject":"Prepare Ussuri M2","status":"MERGED","created":"2020-02-13 06:25:03.000000000","updated":"2020-02-13 17:14:00.000000000","submitted":"2020-02-13 17:14:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707528-1581614040205-a5ae4dff","meta_rev_id":"debb6d166e365d0e9a3a11d1b073f6c871596a7e","_number":707528,"virtual_id_number":707528,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 17:14:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-02-13 17:01:47.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-02-13 17:01:47.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-13 17:01:47.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-02-13 17:14:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8798180b5cc27bcfcffc8c8ecc497becb6b90aae","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-02-13 06:25:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d0735d5de115ac88ad6d62fb146a4c5d8e0cab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 06:39:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/ed5d9b12104c42898ae22c5c77d0be13 : SUCCESS in 5m 34s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/070d646dfc3a4239984863285613fd63 : SUCCESS in 8m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ef34156af0e7fe76d02f120c46dccb726b185dfb","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-13 17:01:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4d784109d9f1dde2e285d226670f41621bc8587c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:01:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9a35b9cb3aa9f84425a78265bc4517fb30bb057","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:14:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"07c95db1e1957dc2788fc3a68e38269e83eff462","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:14:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/bb66c36744ee48a9bcd955de1d09eccd : SUCCESS in 4m 49s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a8145476d44b4b5a9dab4afff0df7a47 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3d4ec70760c5819afe763a377b1cc386e2313a5","revisions":{"f3d4ec70760c5819afe763a377b1cc386e2313a5":{"kind":"REWORK","_number":1,"created":"2020-02-13 06:25:03.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/28/707528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/28/707528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/707528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/707528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/707528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/28/707528/1"}}},"commit":{"parents":[{"commit":"295c4dd6ccbce63553c70be17aff9c310ef57e49","subject":"Fix stdlib version in metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/295c4dd6ccbce63553c70be17aff9c310ef57e49"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-02-13 06:21:28.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-02-13 06:21:28.000000000","tz":480},"subject":"Prepare Ussuri M2","message":"Prepare Ussuri M2\n\nUpdate the version metadata for Ussuri milestone M2\n\nChange-Id: I2cc75c80e2b951bd6ca7b27b507461f8f358c4cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f3d4ec70760c5819afe763a377b1cc386e2313a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f3d4ec70760c5819afe763a377b1cc386e2313a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
