)]}'
{"id":"openstack%2Fpuppet-aodh~714349","triplet_id":"openstack%2Fpuppet-aodh~master~Ie2a6fa3113c1e0a97f73e8e4d750e19003a18f06","project":"openstack/puppet-aodh","branch":"master","topic":"loci_aodh","hashtags":[],"change_id":"Ie2a6fa3113c1e0a97f73e8e4d750e19003a18f06","subject":"Allow wsgi script to be configurable","status":"MERGED","created":"2020-03-23 04:56:36.000000000","updated":"2020-03-27 15:14:35.000000000","submitted":"2020-03-27 15:11:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714349-1585321892331-68d14c44","meta_rev_id":"39fb947cd65cef138cf8eba83b762e3febc1b23d","_number":714349,"virtual_id_number":714349,"owner":{"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-25 21:40:09.000000000","_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-27 15:11:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},{"value":1,"date":"2020-03-25 04:13:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":1,"date":"2020-03-25 06:00:14.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-03-25 17:29:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2020-03-25 19:16:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-03-26 22:23:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-25 04:13:20.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"},{"updated":"2020-03-25 06:00:14.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-03-25 17:29:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-26 22:23:25.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-03-27 15:11:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5718f604a2a705907242c3a3720fe0aa485f492","author":{"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},"date":"2020-03-23 04:56:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3af1a4f1dbd9eddbe2d9319501b0925fa9aeff45","author":{"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},"date":"2020-03-23 05:05:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f9fcf41b207f99fc2d7f5a390ad5c0fcf990a189","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-03-23 05:41:23.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)\n\nLooks good, just some minor stuff around wording","accounts_in_message":[],"_revision_number":2},{"id":"45a4f7708f769769f4f9acda47f853d4d6387545","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 07:11:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c67882f5ebf84c2bb3d3f4b976025593 : SUCCESS in 5m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18aea77790744ce79307c3de05da1bf3 : SUCCESS in 4m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7601a8f625e4c4698322a3056584701 : SUCCESS in 4m 59s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/9a6827a997d24275a89949af984a99de : FAILURE in 15m 36s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4e3dd8e2ca904be4b8ae68834baf5d6b : FAILURE in 13m 19s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/2aef185a8a59491a8481174b09dc047f : FAILURE in 18m 02s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68dace5a0b2f46b994801e0ecb4a7dc6 : FAILURE in 12m 34s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15acf7c579834054afc3e0e6a6c3dd31 : FAILURE in 9m 50s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6a8af488291e4abdaf803417d446c8a2 : SUCCESS in 56m 40s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/9589f5216d9c4e1bb8e35b7d26b62783 : SUCCESS in 1h 46m 41s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2f1f2d54530c48cb9288e078f6cc4014 : SUCCESS in 50m 05s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/04beccb409824015895c04cb410613b3 : SUCCESS in 17m 21s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad6a1fee980f44f4b54e6294ebc1be93 : FAILURE in 6m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfaeacc6b2be44ff881c8c7738e22b06 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"69bd2452641515bb1b5381e6ccdcc6e076e050e4","author":{"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},"date":"2020-03-24 04:28:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2612cf496043bb7d80c4db7c234c2d6cb7c633b8","author":{"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},"date":"2020-03-24 04:48:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ceee1ae92eb15393884ce72c41cef93eeb296250","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 05:45:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7f1377e34094476bba58bf8c7a0f481 : SUCCESS in 4m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4fb4d184630545c9bc4202737b4c0243 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c38e309b0f0042deaebc27553b8bcd9f : SUCCESS in 4m 32s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1d9b98d69777434bb3c2690748c3590c : SUCCESS in 15m 06s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ce35d8d42a70460b861d023e178b5fb1 : SUCCESS in 12m 36s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/7884d4c6cce84d118dcafba9b3f66b7f : SUCCESS in 12m 52s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/523e6a9d09dc420b92e1ea011ac62b01 : SUCCESS in 8m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3736724a9c444817a8773d9151653977 : SUCCESS in 11m 19s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9bf58d2e17ec489cb55d9258bccab1c0 : SUCCESS in 46m 14s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b0dbc0e9fad1483fb173023f0e993f10 : SUCCESS in 51m 53s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1df46f3c062f4530969c163e72968809 : SUCCESS in 17m 23s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ced92d3bdf1f4e60b85a20f8115f8bf8 : FAILURE in 5m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8607e05941424b898618821640c64267 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":4},{"id":"21a9153f63f2fe5e9dae1e63cf0737b551d38c66","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-03-24 22:22:45.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nLooks good, just a minor spelling mistake","accounts_in_message":[],"_revision_number":4},{"id":"cd8611f0a2011288a7fbc59c36127ce769aa71fe","author":{"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},"date":"2020-03-25 00:02:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7e18c8bcfa1f06babbffec31d372fb8b6ebbea34","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-25 00:30:51.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"71ec625bdde08ddf89e907c09d8119e38b896a0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 01:08:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/425f539f01bf456593272f30f8b7d455 : SUCCESS in 5m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ef9571b421348ec9d4eae805df52a5b : SUCCESS in 5m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e74597c1877f4a9b920ac6eb22fb4fd2 : SUCCESS in 5m 34s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7145276538ab4928b3f8ac24b01ba0aa : SUCCESS in 13m 48s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3767a538db764f7ebb045959d65129af : SUCCESS in 10m 22s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/a7f67fd93bde42019cef260e3e38aa92 : SUCCESS in 15m 23s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81187bc026a8436a8f5c7186fa824a61 : SUCCESS in 10m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef8fdc7b426b4b83a7931878a2ebb92a : SUCCESS in 14m 04s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e9dc6f08a15b46229dd50d1623d5d3d6 : SUCCESS in 1h 01m 43s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4a17cffef4ae4f2496908539d81c48be : SUCCESS in 58m 26s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/59e16215c41b4fbf8e217f83cf078ae2 : SUCCESS in 20m 19s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f59c26a631e84d09a2e02160d3ea9764 : FAILURE in 8m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e10d00fcc0140189957ffdac343282f : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":5},{"id":"a98ffd29a363f15d21cad952c9d8e516a6cd9420","author":{"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},"date":"2020-03-25 02:12:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"85913b410ed62220c190f0325dd1f5d34db0d960","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 03:18:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a2fe07ee1cc4cd09ef7c000f61e6442 : SUCCESS in 5m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a475d21fb93c41ed846f78a5d5f2c0d3 : SUCCESS in 5m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f7b1a033bb84033bed9009d44112c7b : SUCCESS in 5m 16s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ae20f10db30e4980bfefcea5ac1a4d91 : SUCCESS in 18m 24s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e8bcbe03302c4ead83392881eadc033c : SUCCESS in 13m 56s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/1b63910b21f64c3f920da5de9d0e163e : SUCCESS in 14m 35s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb954d750c1840cdabf1983bda5279f2 : SUCCESS in 8m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f0c61b7b91c4966ac27ca47c519f806 : SUCCESS in 10m 43s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3f6d559b540848d4bb25e1ab0dc7857a : SUCCESS in 49m 30s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3c852f7151e64c16bc931940db5d750b : SUCCESS in 1h 03m 00s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c96b82d00ab34725811f6619ba7540bf : SUCCESS in 18m 44s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/adb9565480a14ebfb942ecfc1e54d81d : FAILURE in 10m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/beed9ab2c3b747a59d2e9d9e1c60c756 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":6},{"id":"0f16d3a1b423802fe802cfc1f26ead38b12eb268","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-03-25 04:13:20.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"35bc9803efe91cdc71347bdd80c944dcb9b71dd3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-25 06:00:14.000000000","message":"Patch Set 6: Code-Review+1\n\nLGTM, thanks!","accounts_in_message":[],"_revision_number":6},{"id":"9d8b7a3035ece66131419ced86cf8f2e3ab75ea8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-25 17:29:01.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"e010e4d791567b7927b093f2296044606686e35d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-25 19:16:45.000000000","message":"Patch Set 6: Code-Review+2\n\nJust out of curiosity, what do you need it for? :)","accounts_in_message":[],"_revision_number":6},{"id":"7906a3cf63bd3d137612ecadae3358a244f2f75f","author":{"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},"date":"2020-03-25 21:40:09.000000000","message":"Patch Set 6:\n\n\u003e Just out of curiosity, what do you need it for? :)\n\nHi Tobias, we\u0027re looking at using loci containers at the moment so needed to be able to configure this.\n\nThanks for the pluses chaps :)","accounts_in_message":[],"_revision_number":6},{"id":"4647f103415e928516fc456f805d9f604b46dbe5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-26 22:23:25.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"94dfdc9467f47a379e8c39bb8f19e07c2e6ab86e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 22:23:46.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9789622d930a652929b63d640e84c9fc1cd670a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 23:29:48.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c485eb53f2c64dbaa2a7d242bb94ba89 : SUCCESS in 5m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/702e1c9d91e74bb6ab3559fa7db72a87 : SUCCESS in 5m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/988ae2613a99423d895959e23565118d : SUCCESS in 5m 27s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4c57eebb5cf0462890333197cc78b965 : FAILURE in 9m 15s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7800be3e27fa4bed9fc14175b7609ce4 : SUCCESS in 9m 47s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/99d7f90e08e447ffa5dc19775375a01e : SUCCESS in 47m 00s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e19ce26c7f794bb19e3d899097708d16 : SUCCESS in 58m 47s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d7de98aa079041b9819508a249d446fd : FAILURE in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6517914ade7405aa1b7791e58e7f4da : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":6},{"id":"851194e72a210a27daa17fee0862a34545f61249","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-27 13:05:44.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"545a035194bc6f98c5d27b92d4cb77554e9e359f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 14:08:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c0ce206786741d89469b1867e91bd65 : SUCCESS in 4m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a02bd62fa3c41ec8ccfb0f7a36ec83a : SUCCESS in 5m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ed92b123cf646e9b8429c7c4ff8c63a : SUCCESS in 5m 57s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/cb5aea4a43eb47d5a916c96d1e981567 : SUCCESS in 14m 11s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/beb7b513f0154b4d9445723a6e203a7d : SUCCESS in 13m 47s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/f64cb1e98fa14ae0af4e75c7d69e2c38 : SUCCESS in 12m 23s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0b6a841d38c49f88f97eaa6c36adfac : SUCCESS in 10m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a62e6d0c824f4958b2f66b72b375ecd4 : SUCCESS in 11m 18s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/757c4c4f610d4eeaa767062ec490e975 : SUCCESS in 49m 30s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cc887fe8b7bf48a7a0419ba517bfc958 : SUCCESS in 55m 33s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/23bb9de57daa4580ac9a00f152b49aee : SUCCESS in 20m 37s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98067f48be7b45efb91184a00f20f422 : FAILURE in 9m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/842a3c831d9748ec868b3c883d6125e0 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":6},{"id":"b2593cfbdbd23ba02f0e995e333131d1f0ba50cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 14:08:20.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"469f42053e17f0994bce71f1755d0195890d665a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 15:11:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"9ff92446c4315c5af956c8405d2f204941e36592","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 15:11:32.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/926049100e1e4587a8a2924c39e500a4 : SUCCESS in 6m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/085ab77322fa4bdca513cd316a1d14ad : SUCCESS in 7m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f689d2b985a4d7d96857532e4e00aea : SUCCESS in 8m 48s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7ca46ad076d0401eaa4f244b399eae2d : SUCCESS in 22m 27s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83ab05f6086a4912a0dab876318f9510 : SUCCESS in 10m 51s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/82f054a6e4ea46c38e68a6e385c8f34c : SUCCESS in 59m 30s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5b43ab9405434cf1aeaed4af6d5d5e01 : SUCCESS in 1h 00m 03s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2c9c3251d83543e8b0761b77ec6b3e09 : SUCCESS in 19m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a01021f951544b68b624071c342c8052 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":6},{"id":"5f14c73c9f2fb76de3f00323f0913a36ef1abdf0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 15:14:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fc2ec4606c84ca8b3fdfe0e8dfeb6a0 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1c7c146c917c5dbf51db378f8392cc328fe4632b","revisions":{"17babbe69986d15601117d9b22d2ecd3345d17cc":{"kind":"REWORK","_number":1,"created":"2020-03-23 04:56:36.000000000","uploader":{"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},"ref":"refs/changes/49/714349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/49/714349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/1"}}},"commit":{"parents":[{"commit":"dabda383210178653a5189df368423f2c0b918a8","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/dabda383210178653a5189df368423f2c0b918a8"}]}],"author":{"name":"Adrian Smith","email":"aussieade@gmail.com","date":"2020-03-20 04:54:19.000000000","tz":660},"committer":{"name":"Adrian Smith","email":"aussieade@gmail.com","date":"2020-03-23 04:53:53.000000000","tz":660},"subject":"allow wsgi script to be configurable","message":"allow wsgi script to be configurable\n\nChange-Id: Ie2a6fa3113c1e0a97f73e8e4d750e19003a18f06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/17babbe69986d15601117d9b22d2ecd3345d17cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/17babbe69986d15601117d9b22d2ecd3345d17cc"}]},"branch":"refs/heads/master"},"68415ec9cad8a897ea67a14c430716dbe50fbbb7":{"kind":"REWORK","_number":2,"created":"2020-03-23 05:05:49.000000000","uploader":{"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},"ref":"refs/changes/49/714349/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/49/714349/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/2"}}},"commit":{"parents":[{"commit":"dabda383210178653a5189df368423f2c0b918a8","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/dabda383210178653a5189df368423f2c0b918a8"}]}],"author":{"name":"Adrian Smith","email":"aussieade@gmail.com","date":"2020-03-20 04:54:19.000000000","tz":660},"committer":{"name":"Adrian Smith","email":"aussieade@gmail.com","date":"2020-03-23 05:05:39.000000000","tz":660},"subject":"allow wsgi script to be configurable","message":"allow wsgi script to be configurable\n\nChange-Id: Ie2a6fa3113c1e0a97f73e8e4d750e19003a18f06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/68415ec9cad8a897ea67a14c430716dbe50fbbb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/68415ec9cad8a897ea67a14c430716dbe50fbbb7"}]},"branch":"refs/heads/master"},"ce752a1907669c157ab0d4dd306c9abc08e031f8":{"kind":"REWORK","_number":3,"created":"2020-03-24 04:28:34.000000000","uploader":{"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},"ref":"refs/changes/49/714349/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/49/714349/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/3"}}},"commit":{"parents":[{"commit":"dabda383210178653a5189df368423f2c0b918a8","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/dabda383210178653a5189df368423f2c0b918a8"}]}],"author":{"name":"Adrian Smith","email":"aussieade@gmail.com","date":"2020-03-20 04:54:19.000000000","tz":660},"committer":{"name":"Adrian Smith","email":"aussieade@gmail.com","date":"2020-03-24 04:27:41.000000000","tz":660},"subject":"Allow wsgi script to be configurable","message":"Allow wsgi script to be configurable\n\nChange-Id: Ie2a6fa3113c1e0a97f73e8e4d750e19003a18f06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ce752a1907669c157ab0d4dd306c9abc08e031f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ce752a1907669c157ab0d4dd306c9abc08e031f8"}]},"branch":"refs/heads/master"},"ed56152fe1191ad6553637e620797a5d51f88b09":{"kind":"REWORK","_number":4,"created":"2020-03-24 04:48:17.000000000","uploader":{"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},"ref":"refs/changes/49/714349/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/49/714349/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/4"}}},"commit":{"parents":[{"commit":"dabda383210178653a5189df368423f2c0b918a8","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/dabda383210178653a5189df368423f2c0b918a8"}]}],"author":{"name":"Adrian Smith","email":"aussieade@gmail.com","date":"2020-03-20 04:54:19.000000000","tz":660},"committer":{"name":"Adrian Smith","email":"aussieade@gmail.com","date":"2020-03-24 04:48:02.000000000","tz":660},"subject":"Allow wsgi script to be configurable","message":"Allow wsgi script to be configurable\n\nChange-Id: Ie2a6fa3113c1e0a97f73e8e4d750e19003a18f06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ed56152fe1191ad6553637e620797a5d51f88b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ed56152fe1191ad6553637e620797a5d51f88b09"}]},"branch":"refs/heads/master"},"02a3f8bf3348e0c6ef4b64bca1b53c646598ae98":{"kind":"REWORK","_number":5,"created":"2020-03-25 00:02:56.000000000","uploader":{"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},"ref":"refs/changes/49/714349/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/49/714349/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/5"}}},"commit":{"parents":[{"commit":"dabda383210178653a5189df368423f2c0b918a8","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/dabda383210178653a5189df368423f2c0b918a8"}]}],"author":{"name":"Adrian Smith","email":"aussieade@gmail.com","date":"2020-03-20 04:54:19.000000000","tz":660},"committer":{"name":"Adrian Smith","email":"aussieade@gmail.com","date":"2020-03-24 23:44:23.000000000","tz":660},"subject":"Allow wsgi script to be configurable","message":"Allow wsgi script to be configurable\n\nChange-Id: Ie2a6fa3113c1e0a97f73e8e4d750e19003a18f06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/02a3f8bf3348e0c6ef4b64bca1b53c646598ae98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/02a3f8bf3348e0c6ef4b64bca1b53c646598ae98"}]},"branch":"refs/heads/master"},"1c7c146c917c5dbf51db378f8392cc328fe4632b":{"kind":"REWORK","_number":6,"created":"2020-03-25 02:12:09.000000000","uploader":{"_account_id":25635,"name":"Adrian Smith","email":"aussieade@gmail.com","username":"aussieade"},"ref":"refs/changes/49/714349/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/49/714349/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/49/714349/6"}}},"commit":{"parents":[{"commit":"dabda383210178653a5189df368423f2c0b918a8","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/dabda383210178653a5189df368423f2c0b918a8"}]}],"author":{"name":"Adrian Smith","email":"aussieade@gmail.com","date":"2020-03-20 04:54:19.000000000","tz":660},"committer":{"name":"Adrian Smith","email":"aussieade@gmail.com","date":"2020-03-25 02:11:57.000000000","tz":660},"subject":"Allow wsgi script to be configurable","message":"Allow wsgi script to be configurable\n\nChange-Id: Ie2a6fa3113c1e0a97f73e8e4d750e19003a18f06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/1c7c146c917c5dbf51db378f8392cc328fe4632b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/1c7c146c917c5dbf51db378f8392cc328fe4632b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
