)]}'
{"id":"openstack%2Fpuppet-aodh~714881","triplet_id":"openstack%2Fpuppet-aodh~master~Iadbaa88680a15687c1e4af8a28db74a0794531ea","project":"openstack/puppet-aodh","branch":"master","topic":"bug/1868511","hashtags":[],"change_id":"Iadbaa88680a15687c1e4af8a28db74a0794531ea","subject":"Deprecate min_pool_size option","status":"MERGED","created":"2020-03-25 08:23:18.000000000","updated":"2020-03-25 21:17:11.000000000","submitted":"2020-03-25 21:14:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714881-1585170899432-7a5aba1f","meta_rev_id":"cce08ad9bf9aa62cfbf322393054cabb012e9f1a","_number":714881,"virtual_id_number":714881,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-25 21:14:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2020-03-25 17:29:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-03-25 19:12:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-25 17:29:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-25 19:12:09.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-03-25 21:14:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"809bf5af3dd5205c363c6918c32ac4597e0d0a34","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-03-25 08:23:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4871ab5fb95552f465ff05d1aab719bfc1f7030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 09:28:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c5228ede4764425bbc701ed4a926887 : SUCCESS in 6m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fa50e7a9b314e4cba4ada3773999d5b : SUCCESS in 6m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6427f47c91f74dd0952cb578faf9c0c2 : SUCCESS in 6m 24s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e3297e17e54246139678b3b77538fed4 : SUCCESS in 21m 21s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/393ffd53bd924265a9220df98d9cacf1 : SUCCESS in 16m 27s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/f94a62a5ebc4421382d1c5bbf7f70e4e : SUCCESS in 15m 50s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d965b799e52b4ec1942ffe4d5b84e5b3 : SUCCESS in 12m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6eb901ecba96458db0e63cfb37ede15d : SUCCESS in 14m 00s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ba314490d7754e5cb5707ffd92166258 : SUCCESS in 50m 24s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ea470f0bfd554af196cfe5b2a7b05074 : SUCCESS in 55m 52s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2bddc82e7cef4beabd8879586c409ecc : SUCCESS in 18m 06s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/930883283af245238151483a0384062c : FAILURE in 10m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c17c94b361bf411eb1e2897ab97c55a8 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4b63db522196ba9fcac2fb2b5f9188d1ed905465","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-25 17:29:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"71bbb86f5931f3d0bdf8b5cf90fb0687b7d9daaf","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-25 19:12:09.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68516f1e856c30d47a89d7b716ad16dc87b13a27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 19:12:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"515f309f6536e00b752c05f40f6ce363c59d030f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:14:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7e2b30aee464ef7a7ed4e62388e63c6 : SUCCESS in 6m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c76f4cb2dbc47338beb34b090c33c49 : SUCCESS in 5m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c897eef74304b52b5aad7c2398d2c6c : SUCCESS in 5m 55s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4f8086d116bd46ca87de24e304dcc2c1 : SUCCESS in 21m 09s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f018e3e0cfd4b618b87b436341cc506 : SUCCESS in 12m 53s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/22c985aa1004499088613b6a0d038b2e : SUCCESS in 1h 02m 08s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/90f5f13ebb124f43992fe07e6661940e : SUCCESS in 1h 06m 10s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2f7e62a2fb2e4676b4444ebe90864007 : SUCCESS in 20m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/377c1c0cf0c8406eb800fc0cab1f228b : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"7388969e3c4d088cb7d598195b3d5320ab6e8fea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:14:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2ca2466ddead3170ef4824558c01354e7db01ed4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:17:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bfd8ff9e427497581906c7d552fa77a : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea60281eeea5c1d9a4dca907fe23e7880eff29f2","revisions":{"ea60281eeea5c1d9a4dca907fe23e7880eff29f2":{"kind":"REWORK","_number":1,"created":"2020-03-25 08:23:18.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/81/714881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/81/714881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/81/714881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/81/714881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/81/714881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/81/714881/1"}}},"commit":{"parents":[{"commit":"030de21192b2c6251233cb445023c8d52788da09","subject":"Remove idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/030de21192b2c6251233cb445023c8d52788da09"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-03-25 06:53:40.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-03-25 06:53:40.000000000","tz":480},"subject":"Deprecate min_pool_size option","message":"Deprecate min_pool_size option\n\nmin_pool_size option is not used，see:\n\nhttps://review.opendev.org/#/c/565090/\n\nChange-Id: Iadbaa88680a15687c1e4af8a28db74a0794531ea\nCloses-Bug: #1868511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ea60281eeea5c1d9a4dca907fe23e7880eff29f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ea60281eeea5c1d9a4dca907fe23e7880eff29f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
