)]}'
{"id":"openstack%2Fpuppet-aodh~716443","triplet_id":"openstack%2Fpuppet-aodh~master~Iee4deab3b1e71cd2cf2f2efea061db545e9ee47c","project":"openstack/puppet-aodh","branch":"master","topic":"py3","hashtags":[],"change_id":"Iee4deab3b1e71cd2cf2f2efea061db545e9ee47c","subject":"Expect python3-redis in CentOS8 and Fedora","status":"MERGED","created":"2020-03-31 22:38:48.000000000","updated":"2020-04-02 02:48:45.000000000","submitted":"2020-04-02 02:48:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716443-1585795725694-fe09d30a","meta_rev_id":"d9cde57db7f3b099ffee4923b9f1d31eaa8bf624","_number":716443,"virtual_id_number":716443,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 02:48:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-04-02 02:31:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-03-31 23:05:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-04-02 02:31:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 23:05:56.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-02 02:31:02.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-04-02 02:48:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e01e1ed46cc374781be99825f8acec72ba06f1ce","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-31 22:38:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06a744838675815b06c2db6b7080ad14ef6f38d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 22:58:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d87305602b54f89b4a8f84e0a5f3f90 : SUCCESS in 7m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3cd4d40321b4adaa284e36479b7433c : SUCCESS in 6m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4883c73ab955443c9257c149830b6a1f : SUCCESS in 7m 08s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/fe9890a422c14259ad420109265c4e49 : SUCCESS in 17m 22s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/576f102d2cde4175a8df7a1ef30d37b2 : SUCCESS in 17m 04s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/4fe558d6dd324f6889c7751ed92a886a : SUCCESS in 17m 04s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b13b94a064b44f185046e19ebe38a17 : SUCCESS in 13m 29s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0c2ded088b64f74828ac0241ba44875 : SUCCESS in 14m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8379d38a233df5b66a7cf41c357f42feaf3bed89","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-31 23:05:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3951a896999155cef0db5a48fb174d51078ae439","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-02 02:31:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54616d67ef3a4243d26e128917f9e3225679901b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 02:31:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4335c7fe3d98aeeac4bd33eaf0c45d4cb145c681","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 02:48:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4e5146d3df437dbac7e67f6fc92575818e82e770","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 02:48:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/444fe093a3114dee9666b139dca5450f : SUCCESS in 6m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/daee6ed2c136499b9bca85af8a38e5cd : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b96290a2ce444522ad6148aacabf4c69 : SUCCESS in 5m 55s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/1df1f29cc67a4f878b7dd20b1660477c : SUCCESS in 12m 03s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/378355e8fe414c7e9be74e147c9c643a : SUCCESS in 15m 43s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3757a1f5c008477eb0bdb0d3d00eb18f : SUCCESS in 9m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba65e562b91419bc3f0f40932548c4071d5d68e8","revisions":{"ba65e562b91419bc3f0f40932548c4071d5d68e8":{"kind":"REWORK","_number":1,"created":"2020-03-31 22:38:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/716443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/43/716443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/43/716443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/43/716443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/43/716443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/43/716443/1"}}},"commit":{"parents":[{"commit":"c67aeb4b972c58f46cc70d5cf30e5bae98edbff2","subject":"Merge \"Update mysql module to \u003c11.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c67aeb4b972c58f46cc70d5cf30e5bae98edbff2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 22:38:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 22:38:33.000000000","tz":540},"subject":"Expect python3-redis in CentOS8 and Fedora","message":"Expect python3-redis in CentOS8 and Fedora\n\n... because these use python3 instead of python2.\n\nChange-Id: Iee4deab3b1e71cd2cf2f2efea061db545e9ee47c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ba65e562b91419bc3f0f40932548c4071d5d68e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ba65e562b91419bc3f0f40932548c4071d5d68e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
