)]}'
{"id":"openstack%2Fpuppet-aodh~717218","triplet_id":"openstack%2Fpuppet-aodh~stable%2Ftrain~Idb85b81978786be8696715921eb9d95b7d424654","project":"openstack/puppet-aodh","branch":"stable/train","topic":"bug/1850236-stable/train","hashtags":[],"change_id":"Idb85b81978786be8696715921eb9d95b7d424654","subject":"Fix rabbitmq dependency","status":"MERGED","created":"2020-04-03 08:24:40.000000000","updated":"2020-04-03 14:27:46.000000000","submitted":"2020-04-03 14:27:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717218-1585924066489-e8a3316c","meta_rev_id":"7bcd2eb911d7aca000cc847874bf23b6b0060b4c","_number":717218,"virtual_id_number":717218,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 14:27:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-03 14:12:37.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-04-03 08:50:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-04-03 14:12:37.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-04-03 08:24:40.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 08:24:40.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-04-03 08:50:52.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-03 14:12:37.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-03 14:27:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3941a5724a271bb13cc288284337df9156154a36","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-03 08:24:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"148bf9d1c6241ff347bc3d9881ee03f3bde04a4c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-03 08:25:58.000000000","message":"Topic changed from bug/1850236-train to bug/1850236","accounts_in_message":[],"_revision_number":1},{"id":"4fd91cd6ecc8935395d186f0b0b85faeb8d4ac73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 08:30:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/7f16a5e220214461a1dc4f91e721606e : SUCCESS in 3m 47s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d19715dea0c14cd2a36992d3ac90a2c3 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"12326f4a57b6f0318e9504edaa7bd075e7e14862","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-03 08:41:03.000000000","message":"Topic changed from bug/1850236 to bug/1850236-stable/train","accounts_in_message":[],"_revision_number":1},{"id":"f83938a649a1d6280a9609787966e932f848444d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-03 08:50:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"38377d443e9dc9afe314c77400e08f3ee32d6a03","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-03 14:12:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7df243028539fce43ee0e6b94fff2cef9b357034","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 14:12:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d6b851794cbf441da86eb39571f93cf7e580edb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 14:27:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7bcd2eb911d7aca000cc847874bf23b6b0060b4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 14:27:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/c3115ca96609435bb9a936cd58f9b738 : SUCCESS in 3m 57s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/337bb72065194641b8ba5fbdce21342d : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee25acbb405d3ad0a84cb9997b7bc25cb406c6d9","revisions":{"ee25acbb405d3ad0a84cb9997b7bc25cb406c6d9":{"kind":"REWORK","_number":1,"created":"2020-04-03 08:24:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/717218/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/18/717218/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/18/717218/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/18/717218/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/18/717218/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/18/717218/1"}}},"commit":{"parents":[{"commit":"8beb87ab615e8c0962feecda4fb79ba7a2e603a1","subject":"Switch to Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8beb87ab615e8c0962feecda4fb79ba7a2e603a1"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-10-29 01:31:04.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-03 08:24:26.000000000","tz":540},"subject":"Fix rabbitmq dependency","message":"Fix rabbitmq dependency\n\npuppetlabs/rabbitmq [1] is deprecated, and max version is 5.6.0 which can\nnot be fulfilled by this dependency.\n\nCorrect it to puppet-rabbitmq [2] instead which is being actively maintained\nand has the right version.\n\n[1]: https://forge.puppet.com/puppetlabs/rabbitmq\n[2]: https://forge.puppet.com/puppet/rabbitmq\n\nChange-Id: Idb85b81978786be8696715921eb9d95b7d424654\nCloses-Bug: #1850236\n(cherry picked from commit a8fec741ead2f86e0220e2e79fb58a9ea11a37c6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ee25acbb405d3ad0a84cb9997b7bc25cb406c6d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ee25acbb405d3ad0a84cb9997b7bc25cb406c6d9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
