)]}'
{"id":"openstack%2Fpuppet-aodh~718890","triplet_id":"openstack%2Fpuppet-aodh~stable%2Fqueens~I19da3decec3c6867ed817c66c2d600b501e769d3","project":"openstack/puppet-aodh","branch":"stable/queens","hashtags":[],"change_id":"I19da3decec3c6867ed817c66c2d600b501e769d3","subject":"Remove PE requirement from metadata.json","status":"MERGED","created":"2020-04-10 06:51:05.000000000","updated":"2020-04-13 03:52:42.000000000","submitted":"2020-04-13 03:52:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718890-1586749962277-dfe5b5ea","meta_rev_id":"f75ad9356732686f3e60db081c5bda3775b91f38","_number":718890,"virtual_id_number":718890,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-13 03:52:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-13 02:49:45.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-04-13 03:47:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-04-13 03:47:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-10 06:51:05.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-13 02:49:45.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-13 03:47:37.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-04-13 03:52:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2465ccff264b5dba0b562a171ce8ed39a38a30d9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-10 06:51:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96fd8d22c2c01a489ca172f668b484ffcae5ffc1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-10 06:57:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e00be30f9cea96d2ca07b67a7a962da6a9a0f57c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-10 07:07:19.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b3fcfa7403ecd6a59635bfa0612f7e4f6ebf7cb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 07:18:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/6b1890e0846346e8b65678bc121a5660 : SUCCESS in 2m 29s\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/b28428b84c204a22aae5400049a19124 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":3},{"id":"80452a91c9ce62739cd05869bcc03deb104b1978","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-13 02:49:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"122af1c3464b554c73d1b66391e27dcb4bb8f8a7","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-13 03:47:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"33182b272dc78a918bf7c7235594b4d8a1404156","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 03:47:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5bc1477bd9a1df899887b2ec5e332d85e7858524","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 03:52:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f75ad9356732686f3e60db081c5bda3775b91f38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 03:52:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/df65d6ad0ebe420f8208b877fda0c45d : SUCCESS in 3m 26s\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/8f9f8f7cede84c4eb32fbf405ee64314 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3f3a6778bc62ca3f119ebf6d24e6d546f062cc86","revisions":{"c726e6fde04d2660162fdc107df6cfb59d0d328c":{"kind":"REWORK","_number":1,"created":"2020-04-10 06:51:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/718890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/90/718890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/90/718890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/90/718890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/90/718890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/90/718890/1"}}},"commit":{"parents":[{"commit":"dec04b7d08c6e1c80cc7f932e2664b94e0817e5c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/dec04b7d08c6e1c80cc7f932e2664b94e0817e5c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-08-16 21:10:40.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-10 06:50:42.000000000","tz":540},"subject":"Remove PE requirement from metadata.json","message":"Remove PE requirement from metadata.json\n\nThe PE requirement in metadata.json is not\nsupported anymore and puppet-lint will throw:\n\nThe \u0027pe\u0027 requirement is no longer supported by the Forge.\n\nConflicts:\n\tmetadata.json\n\nChange-Id: I19da3decec3c6867ed817c66c2d600b501e769d3\n(cherry picked from commit 659c9b792fc71872bf21f0c28efd817402c8b76a)\n(cherry picked from commit 5eca67fd3faf6d3feed3b05f37191eb4e2ec195b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c726e6fde04d2660162fdc107df6cfb59d0d328c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c726e6fde04d2660162fdc107df6cfb59d0d328c"}]},"branch":"refs/heads/stable/queens"},"278f068cc8f33b18a9ad17302706c2f7e82e6a8e":{"kind":"REWORK","_number":2,"created":"2020-04-10 06:57:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/718890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/90/718890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/90/718890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/90/718890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/90/718890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/90/718890/2"}}},"commit":{"parents":[{"commit":"dec04b7d08c6e1c80cc7f932e2664b94e0817e5c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/dec04b7d08c6e1c80cc7f932e2664b94e0817e5c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-08-16 21:10:40.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-10 06:57:13.000000000","tz":540},"subject":"Remove PE requirement from metadata.json","message":"Remove PE requirement from metadata.json\n\nThe PE requirement in metadata.json is not\nsupported anymore and puppet-lint will throw:\n\nThe \u0027pe\u0027 requirement is no longer supported by the Forge.\n\nConflicts:\n\tmetadata.json\n\nChange-Id: I19da3decec3c6867ed817c66c2d600b501e769d3\n(cherry picked from commit 659c9b792fc71872bf21f0c28efd817402c8b76a)\n(cherry picked from commit 5eca67fd3faf6d3feed3b05f37191eb4e2ec195b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/278f068cc8f33b18a9ad17302706c2f7e82e6a8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/278f068cc8f33b18a9ad17302706c2f7e82e6a8e"}]},"branch":"refs/heads/stable/queens"},"3f3a6778bc62ca3f119ebf6d24e6d546f062cc86":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-10 07:07:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/718890/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/90/718890/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/90/718890/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/90/718890/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/90/718890/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/90/718890/3"}}},"commit":{"parents":[{"commit":"dec04b7d08c6e1c80cc7f932e2664b94e0817e5c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/dec04b7d08c6e1c80cc7f932e2664b94e0817e5c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2018-08-16 21:10:40.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-10 07:07:17.000000000","tz":0},"subject":"Remove PE requirement from metadata.json","message":"Remove PE requirement from metadata.json\n\nThe PE requirement in metadata.json is not\nsupported anymore and puppet-lint will throw:\n\nThe \u0027pe\u0027 requirement is no longer supported by the Forge.\n\nChange-Id: I19da3decec3c6867ed817c66c2d600b501e769d3\n(cherry picked from commit 659c9b792fc71872bf21f0c28efd817402c8b76a)\n(cherry picked from commit 5eca67fd3faf6d3feed3b05f37191eb4e2ec195b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3f3a6778bc62ca3f119ebf6d24e6d546f062cc86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3f3a6778bc62ca3f119ebf6d24e6d546f062cc86"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
