)]}'
{"id":"openstack%2Fpuppet-aodh~720898","triplet_id":"openstack%2Fpuppet-aodh~master~I1e916839f5737a92e62bcc0fc7e0972a66cc7048","project":"openstack/puppet-aodh","branch":"master","topic":"rename-tenant","hashtags":[],"change_id":"I1e916839f5737a92e62bcc0fc7e0972a66cc7048","subject":"Rename tenant to project","status":"MERGED","created":"2020-04-18 15:14:26.000000000","updated":"2020-04-20 20:46:46.000000000","submitted":"2020-04-20 20:43:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720898-1587415414484-354c625b","meta_rev_id":"52422154cef257ebdcd00c01203222b88acef4ef","_number":720898,"virtual_id_number":720898,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-19 22:31:30.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-20 20:43:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-04-20 14:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2020-04-20 15:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-04-20 15:59:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 14:43:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-20 15:59:39.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-20 20:43:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b07cac124594e975349bb55c52ca2f64249856a4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-18 15:14:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb7e3063acf351d3d112a826c06287c1fc61f1de","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-18 15:24:46.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"354409f8d818fefcf4195c5ad4b20ddcd11dc180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 16:00:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf8ea33f246944328e67ec728bee5a6a : SUCCESS in 4m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac30829eff624e31aff228f69c0016e3 : FAILURE in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ecba2bb1ad44ca4ab80b15ef178de41 : FAILURE in 4m 36s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/41630823250144c7a8029ecddbd73cba : FAILURE in 9m 46s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/4b98c14eb05a49e18f0072a4540861d7 : FAILURE in 9m 45s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c60daec8801441f8545d61faf1345e4 : FAILURE in 9m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/038bb6a7d3624367ac1d73bb0ff2f714 : FAILURE in 9m 55s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c35349a74f084d07bd20a6836a01aa1e : FAILURE in 7m 52s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1b419f64980f40ceb55110e3675940c8 : FAILURE in 8m 53s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6d29e41c90394c7f80b1e0b3f1477002 : POST_FAILURE in 16m 32s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ef0fd8e2c91d4733b3b120854ade4991 : FAILURE in 9m 12s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dff3614d021940f5beaf536a357debbf : FAILURE in 6m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a447d7f9972438e84ebba66981e7718 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"711afc0db60ff7a9c09e57ab9a116c486cf84a12","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-19 02:01:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"71a1740b6706f6d32dc1706ae0b560cffb43b36f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 03:04:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eac798dbf0e04bc38cbc8ef5f5ffe690 : SUCCESS in 4m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15cf69523b224529b1c9e89be16d8ac6 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/299e0245915b46999888a58f34d1b6c7 : SUCCESS in 4m 37s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e2b516dafd384938b316a10be96fc75b : SUCCESS in 11m 42s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/00579ba5ff7b4ed18a9bb54a4ec63480 : SUCCESS in 11m 00s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c89c247dd5d945e2bbd6f78db4555392 : SUCCESS in 9m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf0a2c77d7bf4ea9b02432d057aba01c : SUCCESS in 17m 34s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4ba6e0c461914e60affe88def26cb417 : SUCCESS in 1h 01m 47s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4742989dd7654730a11377567edd91a7 : SUCCESS in 54m 03s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/de4812dd312f458292e28423cb3b7d9f : SUCCESS in 17m 14s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/2eb1bac05ba146b4befd8b411bbb8a9e : FAILURE in 8m 34s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec591e992b6147eda652c756f08e6850 : FAILURE in 7m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d73421544d941b1ab32d841eda4bde2 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":3},{"id":"60291961325484b245352c3f42f14ebd8bde7a1d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-19 14:55:26.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"093256b299d2e9cce9202b56f9b2aca9b3d00f88","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-19 14:55:55.000000000","message":"Topic changed from remove-tenant to rename-tenant","accounts_in_message":[],"_revision_number":4},{"id":"4521143b83d6abffe2772f90f6031a64541d542e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 18:59:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2077ea62606745b68b917629d109a6eb : SUCCESS in 4m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c869e9156f3740daa78c06647ca07da0 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7038aa0a52c433fb2a885b721f9e64e : SUCCESS in 4m 29s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/fbbe97e8306247528ec34930ec711ec6 : SUCCESS in 16m 35s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/14ceed87fb014888ae17b6dad5d75bde : SUCCESS in 9m 57s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5578df0f3d7a44ed9f47e06a916543d0 : SUCCESS in 8m 48s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a5192ba24a44e7e946f1a6e2b0eff2d : SUCCESS in 8m 37s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b8043178d23a4088ba924ffeed19e22a : SUCCESS in 43m 14s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0be27a39a2a445388968974409669c91 : SUCCESS in 49m 26s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0cb4b08b4ee14f7bad39dc84e11c3bf3 : TIMED_OUT in 1h 31m 14s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/44f1320b730c4643a4d10d903d589e6b : FAILURE in 7m 29s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/151e1d077370459d943d050012bf5f60 : FAILURE in 7m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d5b4c13e9014a76954a69081367e2d9 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":4},{"id":"5996bd23d3f4882ec20876418c38ab61578f857a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-19 22:31:30.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4daafdf97bf45cee4dd2835b0b3413e128354ccf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 23:37:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99e84cd0901243bf9f64d9e563f70132 : SUCCESS in 4m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e35cfaae92f1475fb11c2fc792754e7c : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd5fee7b4c4942f782973cd7aa24787a : SUCCESS in 4m 04s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/cee5b786607544789dd517017e2f6e9d : SUCCESS in 13m 59s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/32000f1e0cc749569894f2c47d052555 : SUCCESS in 14m 27s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1188f1d86bbd4283872747460ef00ac5 : SUCCESS in 11m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76e84066697943feb8eb5ad580e3c31b : SUCCESS in 8m 49s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/171910177278412080320664e7f0d312 : SUCCESS in 49m 11s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2a7833b1c3cd47a987c1c11e07d80397 : SUCCESS in 1h 00m 33s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d6ec24e3ed0e4de1817d05ed77d21d48 : SUCCESS in 15m 33s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/655a9405b2a2494daced2ce958ac825d : FAILURE in 8m 22s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3086e0c377e48e6adb8ea5437d19dca : FAILURE in 6m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05a66d8f3cfc4df28126c843e10ace2f : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":4},{"id":"f63da99bd0c09b4cecf8a78df8aa9d06f4f42018","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-20 14:43:13.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7d18ad84fe033d02fd770d850a606f8bdd73375f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-20 15:59:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1a3d3807ddcdeefd89664e2837c0f989ba339046","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 19:31:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a1b5d2a029af1ef7de1a3809601a15983a002d48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 20:43:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ce69b1b93862e3f5cdf1351c1ea4e7e74d6a2522","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 20:43:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d333d2701e7a478a8e689db3e42f5623 : SUCCESS in 6m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/edfac75bcd944f23872e59600465e7d8 : SUCCESS in 7m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f8f58ee16d44739bd9cb5b1aa9dd249 : SUCCESS in 6m 58s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/83aa19613e14427fa528f734cafdb952 : SUCCESS in 11m 39s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/8d881d64f5864f439c1f04f44d175f9a : SUCCESS in 14m 31s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b980e8fb3ea4464d801bc4ab30900307 : SUCCESS in 10m 31s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1822b1f26e234311bc876fc2a057448c : SUCCESS in 43m 31s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/58217549ecf540089d1105288f3c1288 : SUCCESS in 1h 02m 17s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c3814e583c4f4863b51b0163f39c1cf0 : SUCCESS in 17m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/745b0d018b6b4518874e7cb2546d5301 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":4},{"id":"52422154cef257ebdcd00c01203222b88acef4ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 20:46:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0496188d2d243349a7bd484f0900310 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b756b9b4c8dea3d75b5ec5d4bec8c653dd047e94","revisions":{"6d4500c980c44ba0e02bc4e62cb41fbc16fb5610":{"kind":"REWORK","_number":1,"created":"2020-04-18 15:14:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/720898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/98/720898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/98/720898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/98/720898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/98/720898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/98/720898/1"}}},"commit":{"parents":[{"commit":"e2d25e0f4ceecd794c31a35f944d5a4f55dd66ea","subject":"Use domain_name instead of domain_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e2d25e0f4ceecd794c31a35f944d5a4f55dd66ea"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-18 15:13:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-18 15:13:17.000000000","tz":540},"subject":"Remove usage of \u0027tenant\u0027","message":"Remove usage of \u0027tenant\u0027\n\n... because in keystone v3 we should use project instead.\n\nChange-Id: I1e916839f5737a92e62bcc0fc7e0972a66cc7048\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6d4500c980c44ba0e02bc4e62cb41fbc16fb5610"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6d4500c980c44ba0e02bc4e62cb41fbc16fb5610"}]},"branch":"refs/heads/master"},"56e232690de701f91eb3dc230ba7de3e7e9d476e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-18 15:24:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/720898/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/98/720898/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/98/720898/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/98/720898/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/98/720898/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/98/720898/2"}}},"commit":{"parents":[{"commit":"ae4390b9edc6548fb8d2d6521f936f7f985c869d","subject":"Use domain_name instead of domain_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ae4390b9edc6548fb8d2d6521f936f7f985c869d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-18 15:13:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-18 15:24:35.000000000","tz":540},"subject":"Remove usage of \u0027tenant\u0027","message":"Remove usage of \u0027tenant\u0027\n\n... because in keystone v3 we should use project instead.\n\nChange-Id: I1e916839f5737a92e62bcc0fc7e0972a66cc7048\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/56e232690de701f91eb3dc230ba7de3e7e9d476e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/56e232690de701f91eb3dc230ba7de3e7e9d476e"}]},"branch":"refs/heads/master"},"01d43eb1afa4e5f596f80d34285c24254b194945":{"kind":"REWORK","_number":3,"created":"2020-04-19 02:01:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/720898/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/98/720898/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/98/720898/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/98/720898/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/98/720898/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/98/720898/3"}}},"commit":{"parents":[{"commit":"ae4390b9edc6548fb8d2d6521f936f7f985c869d","subject":"Use domain_name instead of domain_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ae4390b9edc6548fb8d2d6521f936f7f985c869d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-18 15:13:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-19 02:01:33.000000000","tz":540},"subject":"Remove usage of \u0027tenant\u0027","message":"Remove usage of \u0027tenant\u0027\n\n... because in keystone v3 we should use project instead.\n\nChange-Id: I1e916839f5737a92e62bcc0fc7e0972a66cc7048\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/01d43eb1afa4e5f596f80d34285c24254b194945"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/01d43eb1afa4e5f596f80d34285c24254b194945"}]},"branch":"refs/heads/master"},"b756b9b4c8dea3d75b5ec5d4bec8c653dd047e94":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-04-19 14:55:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/720898/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/98/720898/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/98/720898/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/98/720898/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/98/720898/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/98/720898/4"}}},"commit":{"parents":[{"commit":"ae4390b9edc6548fb8d2d6521f936f7f985c869d","subject":"Use domain_name instead of domain_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ae4390b9edc6548fb8d2d6521f936f7f985c869d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-18 15:13:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-19 14:55:21.000000000","tz":0},"subject":"Rename tenant to project","message":"Rename tenant to project\n\n... because in keystone v3 we use project instead of tenant.\n\nChange-Id: I1e916839f5737a92e62bcc0fc7e0972a66cc7048\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b756b9b4c8dea3d75b5ec5d4bec8c653dd047e94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b756b9b4c8dea3d75b5ec5d4bec8c653dd047e94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
