)]}'
{"id":"openstack%2Fpuppet-aodh~724516","triplet_id":"openstack%2Fpuppet-aodh~master~Idfc00e384be6fff59b4d3cb41024f8dbd83961f8","project":"openstack/puppet-aodh","branch":"master","topic":"update-rabbitmq-version","hashtags":[],"change_id":"Idfc00e384be6fff59b4d3cb41024f8dbd83961f8","subject":"Fix rabbitmq version in metadata","status":"MERGED","created":"2020-04-30 06:05:15.000000000","updated":"2020-04-30 06:37:48.000000000","submitted":"2020-04-30 06:37:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724516-1588228668449-c9a3eab0","meta_rev_id":"0d7de3b471727cabc0ba25e00992848e2b37d5f5","_number":724516,"virtual_id_number":724516,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-30 06:37:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-04-30 06:28:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-04-30 06:28:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 06:37:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34b4d228432d172d3b6d0fd1e5142712ea0f6eef","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-30 06:05:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73e2d751c4db403c7ed481344e5d5d562fc795c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 06:23:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/d16b23fc9dc548358f18cbd0a069906f : SUCCESS in 4m 47s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a773e8e4cd54ee3aad00ccc364b9498 : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e80c307f5a4e5f963db4fc90ae40d759eba3da84","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-30 06:28:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8e39098f5499389b38e9decb6d9bc0e626382e73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 06:28:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f08b85a853fd63183b392e1bad7df13ad71547f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 06:37:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cbc5dd837dc1bf42153be2e13146bfbf184ef80b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 06:37:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/83b60963299240d4a83f22826a82080d : SUCCESS in 4m 03s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecaa60881e824f978c27ae564ecd55c7 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6538ccec9836163b552c4e33063041cd72da0a28","revisions":{"6538ccec9836163b552c4e33063041cd72da0a28":{"kind":"REWORK","_number":1,"created":"2020-04-30 06:05:15.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/16/724516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/16/724516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/16/724516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/16/724516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/16/724516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/16/724516/1"}}},"commit":{"parents":[{"commit":"43b8589a4480ac603aa32852a921a3d65d75ef05","subject":"Change rpsec to rspec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/43b8589a4480ac603aa32852a921a3d65d75ef05"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-04-30 06:04:44.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-04-30 06:04:44.000000000","tz":480},"subject":"Fix rabbitmq version in metadata","message":"Fix rabbitmq version in metadata\n\nrabbitmq is now using the 10.0.1 version, so we should fix the\nrabbitmq version in metadata.\n\nChange-Id: Idfc00e384be6fff59b4d3cb41024f8dbd83961f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6538ccec9836163b552c4e33063041cd72da0a28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6538ccec9836163b552c4e33063041cd72da0a28"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
