)]}'
{"id":"openstack%2Fpuppet-aodh~725121","triplet_id":"openstack%2Fpuppet-aodh~master~I25f0aa5e0b8c8847627e37c396285131df6a2696","project":"openstack/puppet-aodh","branch":"master","topic":"autorequire","hashtags":[],"change_id":"I25f0aa5e0b8c8847627e37c396285131df6a2696","subject":"Use anchor to require necessary packages","status":"MERGED","created":"2020-05-03 15:50:48.000000000","updated":"2020-05-04 01:19:53.000000000","submitted":"2020-05-04 01:19:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725121-1588555193216-2ccb4115","meta_rev_id":"e1f50a81074c99fd9b3be25d2145764c7d19f09d","_number":725121,"virtual_id_number":725121,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-04 01:19:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-04 00:28:39.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-05-03 20:04:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-05-04 00:28:39.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-03 20:04:54.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-04 00:28:39.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-04 01:19:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a92a3e88e537c68b478d05bd32e089699d9d6b5d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-03 15:50:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"226c18b5946ceb1373880af7fd18ce75098a1e68","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-03 15:51:54.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"684cee073e59335a5642fcbf571ad87c99ffd683","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-03 15:52:54.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1355bf597aaa68147887bf9d34c27035bc2cf55c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-03 15:58:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ccb556fabdcd86931ee16075a05f87ea7c057696","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-03 16:18:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"64230c4d01b431371bd2698c76cd18767a3fd4a9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-03 16:29:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1ca46fd5697b62a636b13c4e545760dd1c79fab5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-03 16:41:55.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"c315151d87d7dae87cc335a4a4728694b95a7dbf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-03 16:43:43.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"4f5c6031829e9af80241c5963c825bb3853ca7db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 17:43:03.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1cd22ac570c45158bd4ac518d8daa4f : SUCCESS in 4m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c424aedd6d74dbdad22527560c680f4 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e900ee7bf794033b3381d3e16596448 : SUCCESS in 4m 07s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/06656bbe9e774f7082613e53c6b2711f : SUCCESS in 10m 55s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/68ee2c3e34f94c6da1583826e2ceab93 : SUCCESS in 11m 08s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8f548c584e944a98d10de2c608a055e : SUCCESS in 9m 36s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c0dec77e039452c86b538ca0e4f8902 : SUCCESS in 9m 48s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b32e0322b9924a75848ef0163e9ec907 : SUCCESS in 50m 02s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/356add249ff14ca491e66adfadbe8bc0 : SUCCESS in 54m 29s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/155e535204234b8ba58039c04891ca0b : SUCCESS in 16m 38s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/3eb6b43b811247fa8c0a9f075c72ce66 : FAILURE in 8m 30s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/672fb37781664612ae1651b58f9e475f : FAILURE in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"98c787b7edaf104136f8f7ad3708e388c34b1aa5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-03 20:04:54.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"98dadf2c558a2e67dd345011f8acd8703324aa85","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-04 00:28:39.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"046fc5bad4e0b15c6e3c5b0bce6af053198a0593","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 00:28:50.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"8a89dda89fe225f65c1724a8bd012a885e766f48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 01:19:52.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2de28f3f3bd34ccf96b81c29cf7aba55 : SUCCESS in 5m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/379476c847724beca7a14b41ddcbde1c : SUCCESS in 4m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cc7ea0c9b7c4e4eb775fb5f6369201a : SUCCESS in 4m 04s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ab5d9d956bb948d2ae95df43ce6be557 : SUCCESS in 11m 42s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/51f8aa1540d542b29425dcc2705ef204 : SUCCESS in 15m 15s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7993db1bd1d41b7af44129203b2c714 : SUCCESS in 10m 03s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/26a549cd03c240eeac750eb7f07f35b2 : SUCCESS in 44m 40s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fb8f1aef36d04d468bc3b1577dc485c6 : SUCCESS in 49m 58s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a5c8b774d0c34fb1b9c233b8ee9975e0 : SUCCESS in 25m 06s","accounts_in_message":[],"_revision_number":8},{"id":"e2c85d7d42d8e05328d56a94c667876b1be12065","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 01:19:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"6431321eee91096e58490c611f1f431b15aa3ca9","revisions":{"6752ea627d4b399d682e174d1ab50bb06b4a5d49":{"kind":"REWORK","_number":1,"created":"2020-05-03 15:50:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/725121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/21/725121/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/1"}}},"commit":{"parents":[{"commit":"da1d7ff2c0088af11be30b71f1a4679bedd6e3ca","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/da1d7ff2c0088af11be30b71f1a4679bedd6e3ca"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 15:49:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 15:49:06.000000000","tz":540},"subject":"Use anchor to require necessary packages","message":"Use anchor to require necessary packages\n\n... so correct package are required according to os distro.\n\nChange-Id: I25f0aa5e0b8c8847627e37c396285131df6a2696\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6752ea627d4b399d682e174d1ab50bb06b4a5d49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6752ea627d4b399d682e174d1ab50bb06b4a5d49"}]},"branch":"refs/heads/master"},"32ee63d66ed19c59e60a24b9d876b3db5aaafa2f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-03 15:51:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/725121/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/21/725121/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/2"}}},"commit":{"parents":[{"commit":"da1d7ff2c0088af11be30b71f1a4679bedd6e3ca","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/da1d7ff2c0088af11be30b71f1a4679bedd6e3ca"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 15:49:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 15:51:35.000000000","tz":540},"subject":"Use anchor to require necessary packages","message":"Use anchor to require necessary packages\n\n... so that correct package are required according to os distro.\n\nChange-Id: I25f0aa5e0b8c8847627e37c396285131df6a2696\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/32ee63d66ed19c59e60a24b9d876b3db5aaafa2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/32ee63d66ed19c59e60a24b9d876b3db5aaafa2f"}]},"branch":"refs/heads/master"},"bd0d6bc2ea6f7a46f44f1ed4074d311cc992ded4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-05-03 15:52:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/725121/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/21/725121/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/3"}}},"commit":{"parents":[{"commit":"da1d7ff2c0088af11be30b71f1a4679bedd6e3ca","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/da1d7ff2c0088af11be30b71f1a4679bedd6e3ca"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 15:49:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 15:52:40.000000000","tz":540},"subject":"Use anchor to require necessary packages","message":"Use anchor to require necessary packages\n\n... so that correct package are required according without re-defining\nrequired packages in resource implementations.\n\nChange-Id: I25f0aa5e0b8c8847627e37c396285131df6a2696\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/bd0d6bc2ea6f7a46f44f1ed4074d311cc992ded4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/bd0d6bc2ea6f7a46f44f1ed4074d311cc992ded4"}]},"branch":"refs/heads/master"},"923783b93ef35b3b9cfa4982ced7d4da9a34a637":{"kind":"REWORK","_number":4,"created":"2020-05-03 15:58:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/725121/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/21/725121/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/4"}}},"commit":{"parents":[{"commit":"da1d7ff2c0088af11be30b71f1a4679bedd6e3ca","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/da1d7ff2c0088af11be30b71f1a4679bedd6e3ca"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 15:49:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 15:58:37.000000000","tz":540},"subject":"Use anchor to require necessary packages","message":"Use anchor to require necessary packages\n\n... so that correct package are required according without re-defining\nrequired packages in resource implementations.\n\nChange-Id: I25f0aa5e0b8c8847627e37c396285131df6a2696\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/923783b93ef35b3b9cfa4982ced7d4da9a34a637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/923783b93ef35b3b9cfa4982ced7d4da9a34a637"}]},"branch":"refs/heads/master"},"19c4eca18c756a919e2aee13d8b38ccbc63af5d4":{"kind":"REWORK","_number":5,"created":"2020-05-03 16:18:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/725121/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/21/725121/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/5"}}},"commit":{"parents":[{"commit":"da1d7ff2c0088af11be30b71f1a4679bedd6e3ca","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/da1d7ff2c0088af11be30b71f1a4679bedd6e3ca"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 15:49:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 16:18:01.000000000","tz":540},"subject":"Use anchor to require necessary packages","message":"Use anchor to require necessary packages\n\n... so that correct package are required according without re-defining\nrequired packages in resource implementations.\n\nChange-Id: I25f0aa5e0b8c8847627e37c396285131df6a2696\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/19c4eca18c756a919e2aee13d8b38ccbc63af5d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/19c4eca18c756a919e2aee13d8b38ccbc63af5d4"}]},"branch":"refs/heads/master"},"8d2fd70b69cc8e9f39c809498a956062bdd4454b":{"kind":"REWORK","_number":6,"created":"2020-05-03 16:29:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/725121/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/21/725121/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/6"}}},"commit":{"parents":[{"commit":"da1d7ff2c0088af11be30b71f1a4679bedd6e3ca","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/da1d7ff2c0088af11be30b71f1a4679bedd6e3ca"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 15:49:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 16:28:53.000000000","tz":540},"subject":"Use anchor to require necessary packages","message":"Use anchor to require necessary packages\n\n... so that correct package are required according without re-defining\nrequired packages in resource implementations.\n\nChange-Id: I25f0aa5e0b8c8847627e37c396285131df6a2696\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8d2fd70b69cc8e9f39c809498a956062bdd4454b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8d2fd70b69cc8e9f39c809498a956062bdd4454b"}]},"branch":"refs/heads/master"},"d60df2a8a46c60ed048c1daff50da019b5f7eb65":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-05-03 16:41:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/725121/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/21/725121/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/7"}}},"commit":{"parents":[{"commit":"da1d7ff2c0088af11be30b71f1a4679bedd6e3ca","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/da1d7ff2c0088af11be30b71f1a4679bedd6e3ca"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 15:49:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 16:41:44.000000000","tz":540},"subject":"Use anchor to require necessary packages","message":"Use anchor to require necessary packages\n\n... so that correct packages are required according without re-defining\nthem in resource implementations.\n\nChange-Id: I25f0aa5e0b8c8847627e37c396285131df6a2696\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d60df2a8a46c60ed048c1daff50da019b5f7eb65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d60df2a8a46c60ed048c1daff50da019b5f7eb65"}]},"branch":"refs/heads/master"},"6431321eee91096e58490c611f1f431b15aa3ca9":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2020-05-03 16:43:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/725121/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/21/725121/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/21/725121/8"}}},"commit":{"parents":[{"commit":"da1d7ff2c0088af11be30b71f1a4679bedd6e3ca","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/da1d7ff2c0088af11be30b71f1a4679bedd6e3ca"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 15:49:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 16:43:28.000000000","tz":540},"subject":"Use anchor to require necessary packages","message":"Use anchor to require necessary packages\n\n... so that correct packages are required without re-defining them in\nresource implementations.\n\nChange-Id: I25f0aa5e0b8c8847627e37c396285131df6a2696\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6431321eee91096e58490c611f1f431b15aa3ca9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6431321eee91096e58490c611f1f431b15aa3ca9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
