)]}'
{"id":"openstack%2Fpuppet-aodh~725828","triplet_id":"openstack%2Fpuppet-aodh~stable%2Fussuri~Icf96f0043b954c2be76af1f5ad5ac86d25fe231f","project":"openstack/puppet-aodh","branch":"stable/ussuri","topic":"auth_comment-stable/ussuri","hashtags":[],"change_id":"Icf96f0043b954c2be76af1f5ad5ac86d25fe231f","subject":"Fix incorrect default value in parameter description","status":"MERGED","created":"2020-05-06 10:46:44.000000000","updated":"2020-05-07 15:21:46.000000000","submitted":"2020-05-07 15:21:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725828-1588864906678-d7ad0cb0","meta_rev_id":"292c03323483e279d68c42b2a4c6a952480fe4b0","_number":725828,"virtual_id_number":725828,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 15:21:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-07 14:17:09.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-05-06 14:12:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-05-07 14:17:09.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 14:12:15.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-07 14:17:09.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-07 15:21:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd859930038112bd18c1dadc0cd15e2a7ed044ea","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-06 10:46:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"09949121be54e3bcd64efa1bcf9db4a0afddae1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 11:55:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80b0e906273a48be8c419846ecfb016f : SUCCESS in 5m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70b3f68c71a54170a293febafc9394fc : SUCCESS in 5m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42dfc9a22f744bd29e4973605dfe39a9 : SUCCESS in 5m 19s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4ea71c84e01d4448ae77593791260518 : SUCCESS in 13m 39s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/0bb642fbefc2428797ffca54644235ce : SUCCESS in 13m 29s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ecc8aae06564b179a657b07dd880b11 : SUCCESS in 10m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e21083b0abdc4201a3b3f16660b85691 : SUCCESS in 16m 59s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/55e711cd04e9451a9d29046ae648eeef : SUCCESS in 44m 19s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d5ab16b9993d4160a1b0cdaa22fa9064 : SUCCESS in 55m 00s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4e673f0db9eb468eb85608a3c9ed6a85 : SUCCESS in 18m 51s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b6d9f71fb08440c39e043effe0c769d0 : FAILURE in 8m 58s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d618079ac920440f831dadd0abd391e2 : FAILURE in 6m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5cddb188095303451fd7262bbf1bce110af9de69","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-06 14:12:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"750ff7588ff889d55e361afe54410a9d5f0e5784","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-07 14:17:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a654ae85494cce60f2b84311695a9fc3f6a1e6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:19:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c3ef51cd3232e703919f4ef2ecc368a49254897","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 15:21:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8e434c77d6481569c65859d764e93c1c86d3202a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 15:21:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/684be9b8c4584d77b24b7bb32953d153 : SUCCESS in 8m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a893b05908c41478f62dd4f4c13a075 : SUCCESS in 7m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e2ce1b2a1864846b376a72ca08b76cd : SUCCESS in 7m 07s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e105878734c44c40b56290ccf9ab3e68 : SUCCESS in 17m 20s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/b610281e558b4d2f9c3b32000680c1a6 : SUCCESS in 14m 07s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7da8660bdaf047b1a5b85477dfd788b5 : SUCCESS in 11m 25s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/01c6392b69fc427f89414054a009ec5c : SUCCESS in 51m 01s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f8d906d821df437490eab9cc8fe33ee6 : SUCCESS in 59m 38s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/16112a311bd04345b31d18107589d445 : SUCCESS in 22m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61605fddb395e291535b3c9963894c46cbd665fa","revisions":{"61605fddb395e291535b3c9963894c46cbd665fa":{"kind":"REWORK","_number":1,"created":"2020-05-06 10:46:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/725828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/28/725828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/725828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/725828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/725828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/28/725828/1"}}},"commit":{"parents":[{"commit":"7ac2032e08d3e50e8af5a2155716451e450173a2","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/7ac2032e08d3e50e8af5a2155716451e450173a2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-05 06:32:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-06 10:46:44.000000000","tz":0},"subject":"Fix incorrect default value in parameter description","message":"Fix incorrect default value in parameter description\n\nChange-Id: Icf96f0043b954c2be76af1f5ad5ac86d25fe231f\n(cherry picked from commit 3442aa5842868a2a2313722fb73bf7ac55c84869)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/61605fddb395e291535b3c9963894c46cbd665fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/61605fddb395e291535b3c9963894c46cbd665fa"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
