)]}'
{"id":"openstack%2Fpuppet-aodh~726510","triplet_id":"openstack%2Fpuppet-aodh~master~Ifd726057e513bb3cac3a718f52f7381c97546e7f","project":"openstack/puppet-aodh","branch":"master","topic":"project-ptl-and-contrib-docs","hashtags":[],"change_id":"Ifd726057e513bb3cac3a718f52f7381c97546e7f","subject":"[ussuri][goal] Add CONTRIBUTING.rst","status":"MERGED","created":"2020-05-09 09:25:02.000000000","updated":"2020-05-12 12:43:39.000000000","submitted":"2020-05-12 12:43:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726510-1589287419708-3c493061","meta_rev_id":"82ef93bfefdcff107b1c0b36e0e3d349513d5474","_number":726510,"virtual_id_number":726510,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"date":"2020-05-09 18:28:02.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 12:43:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-05-12 12:26:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-05-11 22:10:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-05-09 23:02:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2020-05-12 12:26:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-09 23:02:42.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"},{"updated":"2020-05-11 22:10:35.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-05-12 12:26:48.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-12 12:43:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ccbd16b5e86fc99138c6545251f3ae4f098b53de","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-09 09:25:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7724c116972432720e09f54b24c8e74771682c6b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-09 09:25:42.000000000","message":"Topic set to project-ptl-and-contrib-docs","accounts_in_message":[],"_revision_number":1},{"id":"9faa6b2a5c56b89de00aa42b3ca935c754a4ee46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 09:32:00.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"aeb9b7e6a81cc73358bd1b6c5aadfa6b0449312c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-09 18:28:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b78ee320a0e447f8a147524cddf61a2f9ec19885","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 19:27:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/723296f35d294b269c1a7eecbd9dbbfe : SUCCESS in 4m 35s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2d55680faeac405b8540c4810ff90ae0 : SUCCESS in 43m 51s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2b7f75269fe443099bf648c5e0fd4484 : SUCCESS in 57m 21s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d8c488bdddb7452786de977daa6b5351 : SUCCESS in 18m 40s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/24331cd9e6704cd8885cfbac9bc8f47d : FAILURE in 7m 59s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f32abfa529b049c1ae85ded1cf93c8aa : FAILURE in 6m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"039cb1f1c74dbaba4e83fdcb8308ae29621e4d68","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-09 23:02:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e62786e3a2a0f67da89b2ffc3fa15005e4708a52","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-11 22:10:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8321ed5f0954246e54c74ceb8cb8786908877866","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-12 12:26:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"99bb58202d7c7b8157de4ed605295e5e00c05b2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 12:27:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6d95220d0da3b11b28400a8091221f8e9cf448f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 12:43:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"82ef93bfefdcff107b1c0b36e0e3d349513d5474","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 12:43:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03ac2e89811e42eb904539c2b744000f : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bee17f0f4d9bf296439e8a3206f941fd50c09833","revisions":{"bee17f0f4d9bf296439e8a3206f941fd50c09833":{"kind":"REWORK","_number":1,"created":"2020-05-09 09:25:02.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/10/726510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/10/726510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/10/726510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/10/726510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/10/726510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/10/726510/1"}}},"commit":{"parents":[{"commit":"3442aa5842868a2a2313722fb73bf7ac55c84869","subject":"Fix incorrect default value in parameter description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3442aa5842868a2a2313722fb73bf7ac55c84869"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-09 09:24:21.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-09 09:24:21.000000000","tz":120},"subject":"[ussuri][goal] Add CONTRIBUTING.rst","message":"[ussuri][goal] Add CONTRIBUTING.rst\n\nStory: 2007236\nTask: 38552\n\nDepends-On: https://review.opendev.org/#/c/726509/\nChange-Id: Ifd726057e513bb3cac3a718f52f7381c97546e7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/bee17f0f4d9bf296439e8a3206f941fd50c09833"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/bee17f0f4d9bf296439e8a3206f941fd50c09833"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
