)]}'
{"id":"openstack%2Fpuppet-aodh~744768","triplet_id":"openstack%2Fpuppet-aodh~master~Ic5055279770342d244ef3a4592067351981b459a","project":"openstack/puppet-aodh","branch":"master","hashtags":[],"change_id":"Ic5055279770342d244ef3a4592067351981b459a","subject":"Expose vhost_custom_fragment","status":"MERGED","created":"2020-08-04 18:07:58.000000000","updated":"2020-08-06 22:04:49.000000000","submitted":"2020-08-06 22:04:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744768-1596751489630-88ec3396","meta_rev_id":"8d1021294cb3255e9a6b0ea3aa72276e1bb11ecc","_number":744768,"virtual_id_number":744768,"owner":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-06 04:13:27.000000000","_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 22:04:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":2,"date":"2020-08-06 05:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-08-06 20:44:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2020-08-06 20:44:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-06 05:48:46.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-08-06 20:44:48.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-06 22:04:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd7535bb1c11646afe718bd385d177afc317e817","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-08-04 18:07:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7370b61e712b642cb81a384d90ff81b14e5526b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 19:15:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2860f1deccbb4370b647444220f295a2 : SUCCESS in 4m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e399344cd4924495b31ba3cc67ec84c4 : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6191011c2c9e4b7fb5945d7fd4a7fbd0 : SUCCESS in 4m 34s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/0ba9d0e0f24241ce81bae84755874a12 : FAILURE in 10m 52s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0205f34e321b4004b7c8858392838eef : FAILURE in 12m 00s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cfa5687f01c4874a44d7779adff7f04 : FAILURE in 8m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8df8255ef8ed41faa528d90507d3b65a : FAILURE in 10m 38s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/db1ff631f3ac45c3bf66cdb9f29c3616 : SUCCESS in 50m 19s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/73f1a86078d74752b35b02a67686c9c2 : SUCCESS in 58m 24s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/48f13e2b609f4469b5ba7a703bcc170f : SUCCESS in 19m 17s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/56d46bffecba4bfbbe59992ede231609 : FAILURE in 9m 26s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4103f9d508ba4af897ce188f62b26f8a : FAILURE in 7m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cacbec89ca624d265db56e5273439df87377b4d0","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-08-05 00:41:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55ce2ddc316d116b93d6faae17eeae906e602d4e","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-08-06 04:12:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b575460ef75eb43a8a8f600dece82c02a7c13584","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-08-06 04:13:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6890637f835bba3e632a32e25032b7317a2012ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 05:32:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/202e41fe18ea4681a6d1c8febf1da4f2 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93acc334aabb479d911d82c639594cd6 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cb9fb6e91c54d08b14979bd301b2809 : SUCCESS in 4m 13s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/73c70548398640a5a45b379b33a85e7a : SUCCESS in 10m 52s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/194aa7c3494640f7a30002a07d67a171 : SUCCESS in 17m 21s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/544eec88d9214e999a555fda627aa2ff : SUCCESS in 14m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e840beb17a8243ca896229ec318eafce : FAILURE in 12m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/221f58a5b64e4b9aa5a2c818ef153e6b : SUCCESS in 36m 56s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5e5b4f906937457bad3f2b608453f43e : SUCCESS in 1h 18m 54s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/84ad6e42bfb648e1ace7f6dca72af41e : SUCCESS in 13m 22s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ed744b39717b46b29b591b3c4cd8cbf0 : FAILURE in 7m 16s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bda682a5277401fb5bc7ad7133b7c3e : FAILURE in 6m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"50507ffea0627ce4c68d7245fe11643dba9c7e1f","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-08-06 05:48:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"22324381990e195682f7e362f211806a185ce4ab","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-06 20:44:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c8e94e715abd52f43eb457b44e7ef246a623ffb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 20:44:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"839ce886f13611a9e1876b47c6454d07d0c66c4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 22:04:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8d1021294cb3255e9a6b0ea3aa72276e1bb11ecc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 22:04:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/581188afb5fe49e580ea46f7f7661e61 : SUCCESS in 4m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2188f2357e2b4255b429535aad589675 : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38ce84adc63f415b99c4746ce22a0638 : SUCCESS in 5m 13s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/06a81ed7115543ee8359fc414facc980 : SUCCESS in 12m 37s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/5c6c939c673b49268dc31f29a674c7e4 : SUCCESS in 13m 33s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b2058150de74917aaac70b0f678cddb : SUCCESS in 13m 35s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ca1dcda70d63460ca1edcee9927859bf : SUCCESS in 51m 20s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1c9a12478a3241619e342149062dd078 : SUCCESS in 1h 17m 03s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5585be5938204dd38fa65cb8dbdc6c4a : SUCCESS in 21m 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"40c6462da5afb8a39590ad9d5d19a9abd2c81cc0","revisions":{"7bf927bccc545bd188dc35cd848df55de62d03f0":{"kind":"REWORK","_number":1,"created":"2020-08-04 18:07:58.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/68/744768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/68/744768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/68/744768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/68/744768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/68/744768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/68/744768/1"}}},"commit":{"parents":[{"commit":"e7c0c5042f82d85bc333b27c1c7b597b5832a245","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e7c0c5042f82d85bc333b27c1c7b597b5832a245"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-04 18:07:12.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-04 18:07:14.000000000","tz":-240},"subject":"Expose vhost_custom_fragment","message":"Expose vhost_custom_fragment\n\nIt is provided by the Puppet class \u0027openstacklib::wsgi::apache\u0027.\nThis change exposes it for the Aodh service.\n\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\nChange-Id: Ic5055279770342d244ef3a4592067351981b459a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/7bf927bccc545bd188dc35cd848df55de62d03f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/7bf927bccc545bd188dc35cd848df55de62d03f0"}]},"branch":"refs/heads/master"},"40c6462da5afb8a39590ad9d5d19a9abd2c81cc0":{"kind":"REWORK","_number":2,"created":"2020-08-06 04:12:46.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/68/744768/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/68/744768/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/68/744768/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/68/744768/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/68/744768/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/68/744768/2"}}},"commit":{"parents":[{"commit":"e7c0c5042f82d85bc333b27c1c7b597b5832a245","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e7c0c5042f82d85bc333b27c1c7b597b5832a245"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-04 18:07:12.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-06 04:12:40.000000000","tz":-240},"subject":"Expose vhost_custom_fragment","message":"Expose vhost_custom_fragment\n\nIt is provided by the Puppet class \u0027openstacklib::wsgi::apache\u0027.\nThis change exposes it for the Aodh service.\n\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\nChange-Id: Ic5055279770342d244ef3a4592067351981b459a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/40c6462da5afb8a39590ad9d5d19a9abd2c81cc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/40c6462da5afb8a39590ad9d5d19a9abd2c81cc0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
