)]}'
{"id":"openstack%2Fpuppet-aodh~745558","triplet_id":"openstack%2Fpuppet-aodh~stable%2Fussuri~Ic5055279770342d244ef3a4592067351981b459a","project":"openstack/puppet-aodh","branch":"stable/ussuri","hashtags":[],"change_id":"Ic5055279770342d244ef3a4592067351981b459a","subject":"Expose vhost_custom_fragment","status":"MERGED","created":"2020-08-10 15:55:31.000000000","updated":"2020-08-10 18:34:47.000000000","submitted":"2020-08-10 18:34:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745558-1597084487208-5080a3ed","meta_rev_id":"824d707512675bc5a55d140da4d668ace49f65cd","_number":745558,"virtual_id_number":745558,"owner":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-10 18:34:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-10 17:50:12.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-10 17:00:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-10 17:50:12.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-10 17:00:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-10 17:50:12.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-10 18:34:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c99d453cc610bd5f2227512c66c078eb9a27a39","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-08-10 15:55:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f059c217980d8b305773c0dfd5ef76578b0f4d7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 16:55:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2bfdbe7e27c143f88cc451a22120ea7d : SUCCESS in 5m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbf496449c6e4201bb870b6b63a6d093 : SUCCESS in 6m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c8f2737413748c4b532831d7d2a9fbe : SUCCESS in 4m 27s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/690f79c6afb243be8e716931a12d70d8 : SUCCESS in 13m 07s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/e4ae5f220ca6441d966004285964997f : SUCCESS in 9m 31s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f919a7ef944a4dbaa94ebc00e55b7848 : SUCCESS in 8m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8160496b880a4ead86cb0bfdc643b181 : FAILURE in 8m 18s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/160f3c89df674fb2aeb660d61c5c1969 : SUCCESS in 53m 43s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/56d9fdae48d54f52b93d5f3d3f80673c : SUCCESS in 56m 52s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ccd75e38aa954034a415c45125b6b10d : SUCCESS in 21m 46s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ff0fb91afd414ce9815872ba9e94ad86 : FAILURE in 12m 03s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7da54e80edd04330bdbd6e63b625851e : FAILURE in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"553c3dee3f2d3bc4f3f1fea6dce886444d001707","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-10 17:00:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b6d8caf1443b382a089be38e60a9cf9832294312","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-10 17:50:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"97c4b43356fb662d340c93099fba7fbddda1b958","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 17:50:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93b4b6ff6e2f1b7becfdc1205f34c8a33c3e246a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 18:34:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e09aa9a3ef434310851b5e21c9ec99bc : SUCCESS in 3m 57s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7e79e09adbb47b7b6323437ec5678c3 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7fbd077cae4419f9fd1b0ab96de70bd : SUCCESS in 5m 49s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b79929c2134544faa05571eee50b185d : SUCCESS in 13m 26s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/d388381ddbd4411abf606c744e821f1b : SUCCESS in 13m 42s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9c5b2c4407a42d8baa1e36f39e97390 : SUCCESS in 9m 02s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ccd6c861c4914d34b1cab83359e9ba68 : SUCCESS in 35m 24s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9a3588b186d841d386150b582597e032 : SUCCESS in 38m 14s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2576fae5abfe4af19089113af68f7da6 : SUCCESS in 21m 11s","accounts_in_message":[],"_revision_number":1},{"id":"824d707512675bc5a55d140da4d668ace49f65cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 18:34:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c2d96fb1f03f520e15bc822fd9ea6379d641086a","revisions":{"c2d96fb1f03f520e15bc822fd9ea6379d641086a":{"kind":"REWORK","_number":1,"created":"2020-08-10 15:55:31.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/58/745558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/58/745558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/745558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/745558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/745558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/58/745558/1"}}},"commit":{"parents":[{"commit":"0cdb00340ac880601592f5437bd9e34c22c03b7c","subject":"Switch puppet-openstack_spec_helper to Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0cdb00340ac880601592f5437bd9e34c22c03b7c"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-04 18:07:12.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-10 15:55:17.000000000","tz":-240},"subject":"Expose vhost_custom_fragment","message":"Expose vhost_custom_fragment\n\nIt is provided by the Puppet class \u0027openstacklib::wsgi::apache\u0027.\nThis change exposes it for the Aodh service.\n\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\nChange-Id: Ic5055279770342d244ef3a4592067351981b459a\n(cherry picked from commit 40c6462da5afb8a39590ad9d5d19a9abd2c81cc0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c2d96fb1f03f520e15bc822fd9ea6379d641086a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c2d96fb1f03f520e15bc822fd9ea6379d641086a"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
