)]}'
{"id":"openstack%2Fpuppet-aodh~745560","triplet_id":"openstack%2Fpuppet-aodh~stable%2Ftrain~Ic5055279770342d244ef3a4592067351981b459a","project":"openstack/puppet-aodh","branch":"stable/train","hashtags":[],"change_id":"Ic5055279770342d244ef3a4592067351981b459a","subject":"Expose vhost_custom_fragment","status":"MERGED","created":"2020-08-10 15:58:35.000000000","updated":"2020-08-12 23:29:54.000000000","submitted":"2020-08-12 23:29:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745560-1597274994507-ea02c699","meta_rev_id":"0a38ef0b0a5a14b976f51a3ee91d8f8f814db808","_number":745560,"virtual_id_number":745560,"owner":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-12 23:29:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-11 19:28:03.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":2,"date":"2020-08-12 21:22:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":1,"date":"2020-08-12 21:22:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-11 19:28:03.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-12 21:22:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-12 23:29:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"905dcca3137461f4b5784b365f24932b61e7a747","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-08-10 15:58:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b90ce05b62b87b40b442c993f7faf561db4ee5b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 16:57:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25a3f9e3edaf42a3ae2fe82471be99c3 : SUCCESS in 4m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2b57fd72a3449f984499cbb57aef952 : SUCCESS in 4m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1790ef21e3f84c898ae2c776eb285342 : SUCCESS in 6m 22s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ecd7b43c113d4dbba509ea340368e151 : FAILURE in 19m 30s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/743b2ef0933f41bc8a99b36c74c79743 : FAILURE in 16m 20s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/c4b62449569d4ed68a1c003d0c0867ad : FAILURE in 11m 37s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ceaa3c369050471a979102818c4f092d : FAILURE in 9m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37966bb1eaf047ea8a4843082d814e0e : FAILURE in 12m 49s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/bdccc3ae3cb14944ad4882d0f28a103c : SUCCESS in 47m 21s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/bd06669a0fb74337a20bac2a219a6d82 : SUCCESS in 52m 35s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/422ffdfc84c345648b6f807cb6522c1f : FAILURE in 11m 35s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/46bfe4bad03e43c7aefabcfd60ffb4b7 : SUCCESS in 18m 33s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a02e38d5bdcf4f0486d1fe8e354f6604 : FAILURE in 17m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7424a2feb25aded0dc66988459e41e1c33c120f","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-08-11 14:19:53.000000000","message":"Patch Set 1:\n\nrecheck The CI failures are with the Apache error_log parameter. Refer to http://paste.openstack.org/show/796737/ from https://708d603f7a9a1a299a2c-c685ac7d1fe5f5adbb4d0f30da5e685e.ssl.cf5.rackcdn.com/745560/1/check/puppet-openstack-unit-5.5-centos-7/ecd7b43/job-output.txt","accounts_in_message":[],"_revision_number":1},{"id":"edd80b884c0471121e0cdcbccd16ce0529725b66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 15:20:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3651c10ca09e4b03aabdb93ad60de26d : SUCCESS in 4m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f91f757e188416fa791de4f5049c437 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e8ff4399fa6400abc63257a91148423 : SUCCESS in 4m 52s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f09221dbb2e345d2b19c34e931228bcf : FAILURE in 21m 08s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/915db17c19c64d229868e25a2f29e4db : FAILURE in 12m 26s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/c48748797a02482aa9f82bfe62d47b74 : FAILURE in 13m 18s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/588a984179294a0ca40c4c3f2ef52044 : FAILURE in 9m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75f33e3395f0444a944e446d5a5cddaa : FAILURE in 9m 23s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c5839d01ca6d443d82c59539db4459e6 : SUCCESS in 47m 54s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/00f35409e34c4b0d9ad430fd2e86e66d : SUCCESS in 55m 58s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c0f0238ad31d47d1bdf30f56e83c7b08 : FAILURE in 12m 39s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1f231fd873fd41cfa139bd7d3bfb4eb6 : SUCCESS in 18m 57s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18204a23176148048662981522297577 : FAILURE in 19m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e3639d5cef70e93c49bb777582de47a60626cd0","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-08-11 16:56:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f08240b642cac47743558dbc8003503d3d96a4fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 18:06:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14d840548cc347e794424a4cb1210304 : SUCCESS in 4m 12s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d0baac4bfb74e31b2103b403ff0ae3b : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36d3f920406b4cad8b04eec701104445 : SUCCESS in 6m 39s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4d29c4a3c3394cafb781831da74921fb : SUCCESS in 12m 45s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a198b95ef7c145bab0116fcbf0cf6cd5 : SUCCESS in 30m 19s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/e50238206bac4f2cb4069279fd0933c2 : SUCCESS in 29m 32s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9cb56f254b54b60851bd8125b7748bd : SUCCESS in 12m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3df6a32da67349fb96300f4a87e9f237 : SUCCESS in 8m 09s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b637da469fce4e86a29e94d8e9665498 : SUCCESS in 46m 49s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/051752fd4aa34426b8951436e90f415d : SUCCESS in 1h 08m 37s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/609c2487eae24cc786cf5573ce974339 : FAILURE in 8m 43s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e1cfcc94b2624745852370d1b3c90c0e : SUCCESS in 16m 41s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65632d7cac5745b5bc5d0b84cd9a7ece : FAILURE in 16m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cb9663b7df60b651be8cb21e7f0ce162db85a108","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-11 19:28:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4269dd3665de2d86464daa0bad4426e27a2cfaab","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-12 21:22:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"235379ee3f1f6fe905024414b5272f2ad8210a5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 21:22:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aa207de7a2e0f19943cfe6e4b99025380fe477b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 23:29:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c6266e36a284abeadb1611b1c9c0f0e : SUCCESS in 4m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4db5166b2fd749dd881a113a2eee56e6 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe4a9dba849940ed9f1c8c523109bc06 : SUCCESS in 4m 14s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/246e144eb5e54a0baea78fc8dbfe652c : SUCCESS in 18m 01s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1cf23bec6f3443aa52825013e0fa12d : SUCCESS in 12m 08s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/394545a4c4de44e0a63bcdb7c83715bc : TIMED_OUT in 2h 05m 05s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/92125a2afeb4496186bf796cb5b8d004 : SUCCESS in 55m 58s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/62cbb042bcb04db0aec23642e5d51350 : SUCCESS in 22m 38s","accounts_in_message":[],"_revision_number":2},{"id":"0a38ef0b0a5a14b976f51a3ee91d8f8f814db808","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 23:29:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff3603ec358097c8fe75a357297a962570edb153","revisions":{"70cd9f5401efe232d55f4b9029d261ecf92e57d8":{"kind":"REWORK","_number":1,"created":"2020-08-10 15:58:35.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/60/745560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/60/745560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/60/745560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/60/745560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/60/745560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/60/745560/1"}}},"commit":{"parents":[{"commit":"ee25acbb405d3ad0a84cb9997b7bc25cb406c6d9","subject":"Fix rabbitmq dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ee25acbb405d3ad0a84cb9997b7bc25cb406c6d9"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-04 18:07:12.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-10 15:57:53.000000000","tz":-240},"subject":"Expose vhost_custom_fragment","message":"Expose vhost_custom_fragment\n\nIt is provided by the Puppet class \u0027openstacklib::wsgi::apache\u0027.\nThis change exposes it for the Aodh service.\n\nConflicts:\n    manifests/wsgi/apache.pp\n    spec/classes/aodh_wsgi_apache_spec.rb\n\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\nChange-Id: Ic5055279770342d244ef3a4592067351981b459a\n(cherry picked from commit 40c6462da5afb8a39590ad9d5d19a9abd2c81cc0)\n(cherry picked from commit c2d96fb1f03f520e15bc822fd9ea6379d641086a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/70cd9f5401efe232d55f4b9029d261ecf92e57d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/70cd9f5401efe232d55f4b9029d261ecf92e57d8"}]},"branch":"refs/heads/stable/train"},"ff3603ec358097c8fe75a357297a962570edb153":{"kind":"REWORK","_number":2,"created":"2020-08-11 16:56:13.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/60/745560/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/60/745560/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/60/745560/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/60/745560/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/60/745560/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/60/745560/2"}}},"commit":{"parents":[{"commit":"ee25acbb405d3ad0a84cb9997b7bc25cb406c6d9","subject":"Fix rabbitmq dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ee25acbb405d3ad0a84cb9997b7bc25cb406c6d9"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-04 18:07:12.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-11 16:56:09.000000000","tz":-240},"subject":"Expose vhost_custom_fragment","message":"Expose vhost_custom_fragment\n\nIt is provided by the Puppet class \u0027openstacklib::wsgi::apache\u0027.\nThis change exposes it for the Aodh service.\n\nConflicts:\n    manifests/wsgi/apache.pp\n    spec/classes/aodh_wsgi_apache_spec.rb\n\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\nChange-Id: Ic5055279770342d244ef3a4592067351981b459a\n(cherry picked from commit 40c6462da5afb8a39590ad9d5d19a9abd2c81cc0)\n(cherry picked from commit c2d96fb1f03f520e15bc822fd9ea6379d641086a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ff3603ec358097c8fe75a357297a962570edb153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ff3603ec358097c8fe75a357297a962570edb153"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
