)]}'
{"id":"openstack%2Fpuppet-aodh~748090","triplet_id":"openstack%2Fpuppet-aodh~master~I12340a02740bc70256953aaf043b567138e89899","project":"openstack/puppet-aodh","branch":"master","topic":"bug/1892952","hashtags":[],"change_id":"I12340a02740bc70256953aaf043b567138e89899","subject":"Add mysql_enable_ndb option","status":"MERGED","created":"2020-08-26 03:54:21.000000000","updated":"2020-08-26 17:16:40.000000000","submitted":"2020-08-26 17:14:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748090-1598462078882-8a9bf4a0","meta_rev_id":"6f1602efb557b594ae0111a352ae7d915cd1c137","_number":748090,"virtual_id_number":748090,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 17:14:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-26 12:47:54.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-26 12:41:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-08-26 15:45:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-26 12:47:54.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-08-26 15:45:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 12:41:33.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-26 12:47:54.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-26 15:45:59.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-26 17:14:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c3a574e3990a147458cc5b193dca77b1ca16c22f","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-08-26 03:54:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f39262afe5b9f0be07b6d664d1c1ff57243e32e9","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-08-26 03:57:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"801b9e1ff9c7026c669086b4c2e3c1734ed12afa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 05:18:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b60ded4a38ed40e9b6c63a491a3e8195 : SUCCESS in 4m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68c52e06439f4bf0bd4b0d6166195176 : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14a8af0ea3e442afa20a812f6b206d35 : SUCCESS in 5m 13s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/00e2033e997547d2a297774ee263063d : SUCCESS in 11m 58s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/74ef2397ec6f431e98c9c81839439166 : SUCCESS in 12m 13s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc7b9d4429bb449b9ff01c865ddcee0e : SUCCESS in 9m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/729a0162c38f4b119551922a77bc4867 : SUCCESS in 12m 31s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/335b87089ef74b6d8987b0c0899523c7 : SUCCESS in 52m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7084971fd23e46c888756b893ac3ed1c : SUCCESS in 1h 13m 04s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/865bc765d55b41bb9040bfd0b9295f66 : FAILURE in 9m 13s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d2945bd21dd4b6788d6424ea0bcb8f0 : FAILURE in 8m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/940458eb9bf0443aa7dd63f470e56721 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"fcc585c871d439020c768a3c701e362132e268ae","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-26 12:41:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"10a9b6c33762c344c44e0fa31757dd78f1b6b20a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-26 12:47:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ef4f829351c9828dcf4846ec33b276d1e685e598","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-26 15:45:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2e4bad7f90b4cc99b3387b370274dce8b1321651","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 15:47:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4582eaeeb415c563baa07b8a19a197bcb34032a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 17:14:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3b0ec22c29b4f24e48c7f9c036cab56582954ef6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 17:14:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e052302f778448d79fc19184ec74688c : SUCCESS in 3m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29792d68320945bda005047f90acb5c9 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da8b4688e41647adb06952f470713828 : SUCCESS in 3m 15s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b8e71dd84ace47c9a7e6fde866066cc5 : SUCCESS in 10m 16s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/76d36ff819484651815f9ccfff173e23 : SUCCESS in 10m 33s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97e58e04b3ae40bfa266ed44ed8eb40e : SUCCESS in 9m 16s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2610f55b5b234c45be2f360cbd983847 : SUCCESS in 46m 34s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f275c80214364692822cd6ef5c875ff8 : SUCCESS in 1h 19m 14s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/d3a68f43442149019138031e51be80cd : SUCCESS in 18m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bffbffb405064ee0884003f0d2b0d8f9 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"3491f8bfec6b76a3040e646ee70c38b3f7dac42b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 17:16:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59b082a7bf094957b7b0dd6856465aa4 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"61df5de0b8796b6f622bbf20eab631834f095bdf","revisions":{"62603b4f78d381ba7859652ac3b55ea8e4930e2c":{"kind":"REWORK","_number":1,"created":"2020-08-26 03:54:21.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/90/748090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/90/748090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/90/748090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/90/748090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/90/748090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/90/748090/1"}}},"commit":{"parents":[{"commit":"89da4b14fa7512a9cc0afd9748c5b1c539834091","subject":"Add support for aodh-expirer cron job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/89da4b14fa7512a9cc0afd9748c5b1c539834091"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-26 03:53:17.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-26 03:53:17.000000000","tz":480},"subject":"Add mysql_enable_ndb option","message":"Add mysql_enable_ndb option\n\nAdd mysql_enable_ndb parameter to select mysql storage engine.\n\nChange-Id: I12340a02740bc70256953aaf043b567138e89899\nCloses-Bug: #1892952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/62603b4f78d381ba7859652ac3b55ea8e4930e2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/62603b4f78d381ba7859652ac3b55ea8e4930e2c"}]},"branch":"refs/heads/master"},"61df5de0b8796b6f622bbf20eab631834f095bdf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-26 03:57:22.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/90/748090/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/90/748090/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/90/748090/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/90/748090/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/90/748090/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/90/748090/2"}}},"commit":{"parents":[{"commit":"89da4b14fa7512a9cc0afd9748c5b1c539834091","subject":"Add support for aodh-expirer cron job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/89da4b14fa7512a9cc0afd9748c5b1c539834091"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-26 03:53:17.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-26 03:55:25.000000000","tz":480},"subject":"Add mysql_enable_ndb option","message":"Add mysql_enable_ndb option\n\nAdd mysql_enable_ndb parameter to select mysql storage engine.\n\nChange-Id: I12340a02740bc70256953aaf043b567138e89899\nDepends-On: https://review.opendev.org/#/c/748067\nCloses-Bug: #1892952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/61df5de0b8796b6f622bbf20eab631834f095bdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/61df5de0b8796b6f622bbf20eab631834f095bdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
