)]}'
{"id":"openstack%2Fpuppet-aodh~748171","triplet_id":"openstack%2Fpuppet-aodh~stable%2Fussuri~I272df431b8e71a004761e6a61b3fec9a915b4560","project":"openstack/puppet-aodh","branch":"stable/ussuri","topic":"nodepool-centos8-stable/ussuri","hashtags":[],"change_id":"I272df431b8e71a004761e6a61b3fec9a915b4560","subject":"Add CentOS8 to nodesets","status":"MERGED","created":"2020-08-26 12:19:22.000000000","updated":"2020-08-26 13:16:50.000000000","submitted":"2020-08-26 13:16:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748171-1598447810112-bc9ee2cc","meta_rev_id":"3d8d9a26a2775b8961a88876e3b8fdb5853d0662","_number":748171,"virtual_id_number":748171,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 13:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-26 12:53:45.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-26 12:39:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-26 12:53:45.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 12:39:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-26 12:53:45.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-26 13:16:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53d23fa7ed3772245a047c11094a595a27b46de8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-26 12:19:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5fb89771e98c3feda89f5b678b9121b23a368c1d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-26 12:39:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"37c4dc954bc37ab5750b60d23d3bb470ef30b1b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 12:49:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5fe1967323743358d82b03e70a7633d : SUCCESS in 4m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7cf949591c946aeb109e43203bb2cb3 : SUCCESS in 4m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c380426fd364285adb3437d224b2eae : SUCCESS in 4m 26s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fa7f20b94d1c44cf811e77dd83cdb3db : SUCCESS in 20m 13s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/0b2a0988918b439e9e9b70e292c41a1f : FAILURE in 8m 08s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3780592dbcef41f9a9d6ac13208fda2a : FAILURE in 7m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"958186c2fb309bf9481384f69879f17fe553b554","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-26 12:53:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8df171a0a77f6b75815e80e7c6e76b285b63d586","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 12:53:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce87bf72b0291729e11bb41a8f180e7bccf8fa30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 13:16:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c38728c27b449798d2c0284992d0eb8 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d688e9a995545938788a71d0d2ed014 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e748a2f0f6c4365b3aa3b8472414edd : SUCCESS in 4m 33s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0c18a51af0bf415488f0370e3c02231a : SUCCESS in 17m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b879511481285a8f7d14d2aaae349d0ee4d7d7ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 13:16:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c835f3f26e5b46aa20ca14bdc3de4c5c2b60902","revisions":{"8c835f3f26e5b46aa20ca14bdc3de4c5c2b60902":{"kind":"REWORK","_number":1,"created":"2020-08-26 12:19:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/748171/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/71/748171/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/71/748171/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/71/748171/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/71/748171/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/71/748171/1"}}},"commit":{"parents":[{"commit":"2d840a5c9e5e67543fedb4437b142c41e7e3d5ff","subject":"Use common class to manage apache service in beaker job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/2d840a5c9e5e67543fedb4437b142c41e7e3d5ff"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-10 12:40:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-26 12:19:22.000000000","tz":0},"subject":"Add CentOS8 to nodesets","message":"Add CentOS8 to nodesets\n\nChange-Id: I272df431b8e71a004761e6a61b3fec9a915b4560\n(cherry picked from commit 7e5603773e7bb113d44dd0ddd9ad75c71fc1404a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8c835f3f26e5b46aa20ca14bdc3de4c5c2b60902"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8c835f3f26e5b46aa20ca14bdc3de4c5c2b60902"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
