)]}'
{"id":"openstack%2Fpuppet-aodh~749097","triplet_id":"openstack%2Fpuppet-aodh~master~Ib9b27ca2bb6b0a570b0d41ece72fd470294067ea","project":"openstack/puppet-aodh","branch":"master","hashtags":[],"change_id":"Ib9b27ca2bb6b0a570b0d41ece72fd470294067ea","subject":"Add Puppet Litmus","status":"MERGED","created":"2020-08-31 18:36:05.000000000","updated":"2020-09-10 02:05:31.000000000","submitted":"2020-09-10 02:05:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749097-1599703531646-210ac784","meta_rev_id":"3c46d1bfd6cfce47491eb9414fdfe3604827decc","_number":749097,"virtual_id_number":749097,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2020-09-09 22:55:01.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 02:05:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-08 18:27:18.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-08 18:18:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-08 18:27:18.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-09-09 21:28:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 18:27:18.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-09 21:28:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-09-10 02:05:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e091eda0fe5a9d0d52ed20d4444218783e824128","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 18:36:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56aed70ec752a220512eedf69bde0cd1d7fa7fc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 19:39:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fc9fe0e5e8f44d197bd099b05df5fa4 : SUCCESS in 2m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1b7d8d09f204746a8c5fc14a185ef31 : SUCCESS in 2m 51s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/bb826723e60148d098958a754027176b : SUCCESS in 9m 58s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c54f26fe4e7548f9a138e8b168174240 : SUCCESS in 8m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38d9afdf38f242379bb2e1e798d26d2b : SUCCESS in 8m 20s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8bcf5b4974bf4347ba9574a4a6690a78 : SUCCESS in 54m 49s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/06ad6a4e75fa421ca27f73aee63f8704 : SUCCESS in 53m 48s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/bafbe4da185d4276917ebd06b407f170 : SUCCESS in 7m 34s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8bc54f3620041c5aa14402728382b9a : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2933dabb1e00ebd2aad6bf8210fef202839f341c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-08 18:18:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6729542d64960853fe7cb7f185de76c0aa8b0a7d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-08 18:27:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bc56c0725b1f6d2708ece3ca9171dedd24ed4655","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-09 21:28:12.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f00beeb10540543972f33e60003269a5eeff0b90","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-09 22:55:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4629886e18a35375b085a94235dbfa5e94ae7a34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 00:30:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e81add4483e45619182e15d1009f807 : SUCCESS in 3m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c8492fa9d4d4806bd7b75240fae298a : SUCCESS in 4m 22s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1a6e3631b9c6438aa89eeb3c16c02396 : SUCCESS in 10m 40s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c81419b33cf4aa2863134925fbe6f79 : SUCCESS in 8m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da8a2ec6b60f487b9cb1c7c8a63cbbf0 : SUCCESS in 9m 14s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d3f41a3e0c4e41f8b12289cc690f3b8b : SUCCESS in 54m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/03f59dda9e6e4b92a254a470e68e99e0 : SUCCESS in 1h 14m 59s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/5c0700a3226a4cc882a6656ddb68811c : SUCCESS in 18m 11s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e22ac61e565d416b9002e8c2bfd079d5 : SUCCESS in 13m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9d41c09b549f588769a6e1895e1ac656b7929d22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 00:31:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"426dc8aa6a503b114568cdc97591fa5d1d8c5278","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 02:05:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3c46d1bfd6cfce47491eb9414fdfe3604827decc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 02:05:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/acf5fe78a9124e43a2715ddb36e413c6 : SUCCESS in 4m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8ee87b76310411fb58f640b4d748bb1 : SUCCESS in 4m 07s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b294a176125f4830827dc68fad2fd651 : SUCCESS in 14m 10s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e853807a55cb41268d83259809ce5c3c : SUCCESS in 7m 07s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3f40b1de21fe475383dd7b0c68dfdc9e : SUCCESS in 49m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bcb7c01d67284d36a4d05553c3f127b4 : SUCCESS in 1h 25m 16s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/86f4c5b827a14b5ea1fe8445e32c9e45 : SUCCESS in 19m 41s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c19a350868841109449695d09a8d59d : SUCCESS in 18m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e4d683373b842740edc4321a81bf3449e4b0991","revisions":{"4e4d683373b842740edc4321a81bf3449e4b0991":{"kind":"REWORK","_number":1,"created":"2020-08-31 18:36:05.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/97/749097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/97/749097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/97/749097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/97/749097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/97/749097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/97/749097/1"}}},"commit":{"parents":[{"commit":"61df5de0b8796b6f622bbf20eab631834f095bdf","subject":"Add mysql_enable_ndb option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/61df5de0b8796b6f622bbf20eab631834f095bdf"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 18:35:48.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 18:35:48.000000000","tz":120},"subject":"Add Puppet Litmus","message":"Add Puppet Litmus\n\nDepends-On: https://review.opendev.org/#/c/740601/\nChange-Id: Ib9b27ca2bb6b0a570b0d41ece72fd470294067ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/4e4d683373b842740edc4321a81bf3449e4b0991"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/4e4d683373b842740edc4321a81bf3449e4b0991"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
