)]}'
{"id":"openstack%2Fpuppet-aodh~754400","triplet_id":"openstack%2Fpuppet-aodh~master~Ie93782d2d54446d1ba5ff59ad6f1843006042f79","project":"openstack/puppet-aodh","branch":"master","topic":"rspec-puppet","hashtags":[],"change_id":"Ie93782d2d54446d1ba5ff59ad6f1843006042f79","subject":"DNM: Testing possible fix","status":"ABANDONED","created":"2020-09-25 14:10:42.000000000","updated":"2020-09-26 00:05:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"20e53a055f47024bb0f7a99354d27cb37a2a35e5","_number":754400,"virtual_id_number":754400,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-25 16:00:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-25 16:00:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5799d45031863e27245683622e4232c5f8cc995b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-25 14:10:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"304f88d96fc74ff8c65d8f3590171d4e5de93b7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 14:30:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec0420f140994e7fb5c27865341cd80a : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc56190900e7469e81d1282d99cb0c98 : SUCCESS in 4m 54s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/a5205bc848d84064b45515d68a638a0f : FAILURE in 10m 19s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/993ae8360b164fdaa0db28871c46d823 : FAILURE in 8m 42s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/423b17128a2a424793a28b2b8825837a : FAILURE in 12m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9322c6b4cb6248de94114bf47a4e4241 : FAILURE in 5m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/45b32e021aa740a5804b64fcecf030e2 : FAILURE in 10m 02s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/b4960b9e979b41489c8517f3d5b886f8 : FAILURE in 11m 29s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13252f97b2d14ed5bd43b40cbacd29ba : FAILURE in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7330ed6af456d015803799bd1e8a5d63da53de6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-25 14:40:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"38f97ce5b2e39af8d40756860c7fe85566b09ee3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 16:00:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2756dfd98dab4b9795af23ba1e4e071e : SUCCESS in 5m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2602c5bab6f34a03ad9c3a46d0cc6c36 : SUCCESS in 3m 09s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/06c3fa5a53d84fd7b9e2e25934dfd3fe : FAILURE in 9m 51s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5e8cb288f44492eb93f06111744238a : FAILURE in 9m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf6cdbe4456f4cdeb834609d9579de48 : FAILURE in 9m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/875abb372130490a996208c4e80b714e : SUCCESS in 50m 29s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/60630cbe4731452bab36cd4ff74f4846 : SUCCESS in 1h 08m 06s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/26a1c4169667494795776d80eeea2a38 : SUCCESS in 17m 26s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c00ca78b1d7a430f8861fcc7b86dc611 : SUCCESS in 14m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"31d821a65151e5e4af8d21ed0cb7f6a6a6d51e4d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-26 00:05:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0a6a08a14d1f8104a8b6af57ac73bca98e4010fc","revisions":{"1d2677122c55196baa6129f1c4a400e32a85e7a5":{"kind":"REWORK","_number":1,"created":"2020-09-25 14:10:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/754400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/00/754400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/00/754400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/00/754400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/00/754400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/00/754400/1"}}},"commit":{"parents":[{"commit":"a9f5d247b23dfd07e56ac15bc242a1d16dd2461c","subject":"Fix rspec-puppet 2.7.1 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/a9f5d247b23dfd07e56ac15bc242a1d16dd2461c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-25 14:10:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-25 14:10:27.000000000","tz":540},"subject":"DNM: Testing possible fix","message":"DNM: Testing possible fix\n\nChange-Id: Ie93782d2d54446d1ba5ff59ad6f1843006042f79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/1d2677122c55196baa6129f1c4a400e32a85e7a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/1d2677122c55196baa6129f1c4a400e32a85e7a5"}]},"branch":"refs/heads/master"},"0a6a08a14d1f8104a8b6af57ac73bca98e4010fc":{"kind":"REWORK","_number":2,"created":"2020-09-25 14:40:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/754400/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/00/754400/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/00/754400/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/00/754400/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/00/754400/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/00/754400/2"}}},"commit":{"parents":[{"commit":"a9f5d247b23dfd07e56ac15bc242a1d16dd2461c","subject":"Fix rspec-puppet 2.7.1 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/a9f5d247b23dfd07e56ac15bc242a1d16dd2461c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-25 14:10:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-25 14:40:49.000000000","tz":540},"subject":"DNM: Testing possible fix","message":"DNM: Testing possible fix\n\nChange-Id: Ie93782d2d54446d1ba5ff59ad6f1843006042f79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0a6a08a14d1f8104a8b6af57ac73bca98e4010fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0a6a08a14d1f8104a8b6af57ac73bca98e4010fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
