)]}'
{"id":"openstack%2Fpuppet-aodh~755140","triplet_id":"openstack%2Fpuppet-aodh~master~I0db901e53ec53d8a5098e7f7aeb8cda3c52c79a0","project":"openstack/puppet-aodh","branch":"master","topic":"default-not-undef","hashtags":[],"change_id":"I0db901e53ec53d8a5098e7f7aeb8cda3c52c79a0","subject":"Remove redundant usage of undef","status":"MERGED","created":"2020-09-30 00:06:29.000000000","updated":"2020-09-30 16:21:26.000000000","submitted":"2020-09-30 16:20:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755140-1601482801934-e8c133ac","meta_rev_id":"e37bd5ccc1bb563e91bd034aabaec68916a68af9","_number":755140,"virtual_id_number":755140,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-30 16:20:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-30 14:38:13.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-30 12:14:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-30 14:38:13.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 12:14:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-09-30 14:38:13.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-30 16:20:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54343b3a60ca51fd194f696fddb0d4088c58bca3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-30 00:06:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"348e7f45d37ead43bdc609c66bd9f5ecb3ce0c7d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-30 00:23:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"51470afdef0c1f20564892561a40f385f78743bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 01:50:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10b6d6325f1a44ea9297c19f752acd8c : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e28d0a561d44a24a2fa0dfd3eb7c6a0 : SUCCESS in 3m 29s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d8a3ea57e8a24b989760e38f73bf3754 : SUCCESS in 15m 03s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e3b1ab7544e493383edb8453730f1e0 : SUCCESS in 12m 09s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2059a1eaaf76420ca995b968baad5f1b : SUCCESS in 13m 33s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4c572be46c4e4d4db5d4cbc088dd475a : SUCCESS in 1h 01m 56s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cb26020e00f74cf3b3dbce8b55d28524 : SUCCESS in 1h 22m 49s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d0beed32eba44303a4313c5a05abd311 : SUCCESS in 25m 41s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93a97dd311374d218cd6acece5587dd4 : SUCCESS in 19m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/735a168774244a44b8e21b8fcadbd7b7 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":2},{"id":"35298ee446f24a8794d079f984f0aaf83a92f7e0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-30 10:07:58.000000000","message":"Patch Set 2: Code-Review-1\n\nCould you add a TODO note so that we remember to remove the pick()\u0027s after one cycle?","accounts_in_message":[],"_revision_number":2},{"id":"f7b41c30ce324d8d0593497cd65da4634dc559b0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-30 11:56:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"17f42432ab51a08eb7237b8af4004a13b04f881f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-30 12:14:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"15949e1c4290aa1ef8f928bb9cefb5801abf631a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 13:06:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4dde776adc0445809a624a1d4fc48248 : SUCCESS in 3m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/467900609f9c4c3b9f5b5ec3d2dd5321 : SUCCESS in 3m 17s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9f7a90cf80dd469a978c9e260fc94412 : SUCCESS in 13m 48s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d3d23ddb76d410c893487f86c1a60cf : SUCCESS in 7m 11s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7eb0e437530a4107b35d93a8a551c961 : SUCCESS in 8m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/cfbad928253e401d921cb4a664e3fe88 : SUCCESS in 49m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/01c7e28e0c884f1298f9cc2d34c03d0c : SUCCESS in 1h 07m 27s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/4e95a2f179a047038bd3dd25f6af8f5e : SUCCESS in 19m 14s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c20e1c19187b4a1a9c4bf258f063406d : SUCCESS in 14m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21fca270687d4a08b30ed9542dd77e02 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":3},{"id":"ba328b7179d28aab7d080c3b6c9a083cc442f61e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-30 14:38:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4f067f2466fbada6c5e52737d75b5f6f4f83ff65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 14:38:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f570714da442fa7a5fec783aa6701df1c8ca5f63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 16:20:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b066e1392a5b84a489d96219c75452083b6b0c4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 16:20:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5bdab517fa9e46de9c0e903f297e0ff1 : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c245759a8fa347c99af525c1599b3cff : SUCCESS in 3m 06s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/7571db4ad81d41adbf598152916e8bf4 : SUCCESS in 10m 49s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69db91d6173f4deab867cb891fe708c4 : SUCCESS in 8m 14s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/127fe403ec0a4e39916ac7406715d4ef : SUCCESS in 52m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ab060558b8a640fcbde0e1b0dbc7f476 : SUCCESS in 1h 31m 13s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/6fee4b6e2a0b45939d0e6e3a1a499660 : SUCCESS in 18m 37s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5647cda1bed48dead6d4398558bdc50 : SUCCESS in 22m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a6c500b218840d1b24d4e836b7cbd16 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"b5446581b3fd8eb2fb44d72175cfaa58e3948421","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 16:21:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63065cc4210c4108aef45e20a0d10cd3 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fa753eafe840ad1582e48442a975629e8694a058","revisions":{"456b49b00af2853b9cb573498295c0d97aaf1c26":{"kind":"REWORK","_number":1,"created":"2020-09-30 00:06:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/755140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/40/755140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/755140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/755140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/755140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/40/755140/1"}}},"commit":{"parents":[{"commit":"61df5de0b8796b6f622bbf20eab631834f095bdf","subject":"Add mysql_enable_ndb option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/61df5de0b8796b6f622bbf20eab631834f095bdf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-30 00:00:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-30 00:01:49.000000000","tz":540},"subject":"Remove redundant usage of undef","message":"Remove redundant usage of undef\n\nThe parameters added by 1c7c146c917c5dbf51db378f8392cc328fe4632b use\nundef as the default values and automatically replaces undef by the\nactual value expected by distro, but this is redundant behavior.\nThis patch makes actual values used as the defaults to make logic to\npick up the value more simple.\n\nNote that previous behavior with undef is still kept to keep backward\ncompatibility during this cycle, but it will be removed in a future\nrelease.\n\nChange-Id: I0db901e53ec53d8a5098e7f7aeb8cda3c52c79a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/456b49b00af2853b9cb573498295c0d97aaf1c26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/456b49b00af2853b9cb573498295c0d97aaf1c26"}]},"branch":"refs/heads/master"},"78e776882ca5fa76fd08fba1e9dfb872f1c6ef68":{"kind":"REWORK","_number":2,"created":"2020-09-30 00:23:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/755140/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/40/755140/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/755140/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/755140/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/755140/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/40/755140/2"}}},"commit":{"parents":[{"commit":"61df5de0b8796b6f622bbf20eab631834f095bdf","subject":"Add mysql_enable_ndb option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/61df5de0b8796b6f622bbf20eab631834f095bdf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-30 00:00:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-30 00:23:42.000000000","tz":540},"subject":"Remove redundant usage of undef","message":"Remove redundant usage of undef\n\nThe parameters added by 1c7c146c917c5dbf51db378f8392cc328fe4632b use\nundef as the default values and automatically replaces undef by the\nactual value expected by distro, but this is redundant behavior.\nThis patch makes actual values used as the defaults to make logic to\npick up the value more simple.\n\nNote that previous behavior with undef is still kept to keep backward\ncompatibility during this cycle, but it will be removed in a future\nrelease.\n\nChange-Id: I0db901e53ec53d8a5098e7f7aeb8cda3c52c79a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/78e776882ca5fa76fd08fba1e9dfb872f1c6ef68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/78e776882ca5fa76fd08fba1e9dfb872f1c6ef68"}]},"branch":"refs/heads/master"},"fa753eafe840ad1582e48442a975629e8694a058":{"kind":"REWORK","_number":3,"created":"2020-09-30 11:56:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/755140/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/40/755140/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/755140/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/755140/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/755140/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/40/755140/3"}}},"commit":{"parents":[{"commit":"61df5de0b8796b6f622bbf20eab631834f095bdf","subject":"Add mysql_enable_ndb option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/61df5de0b8796b6f622bbf20eab631834f095bdf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-30 00:00:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-30 11:55:59.000000000","tz":540},"subject":"Remove redundant usage of undef","message":"Remove redundant usage of undef\n\nThe parameters added by 1c7c146c917c5dbf51db378f8392cc328fe4632b use\nundef as the default values and automatically replaces undef by the\nactual value expected by distro, but this is redundant behavior.\nThis patch makes actual values used as the defaults to make logic to\npick up the value more simple.\n\nNote that previous behavior with undef is still kept to keep backward\ncompatibility during this cycle, but it will be removed in a future\nrelease.\n\nChange-Id: I0db901e53ec53d8a5098e7f7aeb8cda3c52c79a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/fa753eafe840ad1582e48442a975629e8694a058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/fa753eafe840ad1582e48442a975629e8694a058"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
