)]}'
{"id":"openstack%2Fpuppet-aodh~755932","triplet_id":"openstack%2Fpuppet-aodh~master~Ie3111d5a168e8d15f3e04d410a5a5c30b44324c2","project":"openstack/puppet-aodh","branch":"master","topic":"expirer-package","hashtags":[],"change_id":"Ie3111d5a168e8d15f3e04d410a5a5c30b44324c2","subject":"Manage aodh-expirer package by aodh::expirer class","status":"MERGED","created":"2020-10-03 14:05:25.000000000","updated":"2020-10-05 14:54:46.000000000","submitted":"2020-10-05 14:53:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755932-1601909588666-48b48ccc","meta_rev_id":"9b703f0d42296291fdc839802c7339927724118c","_number":755932,"virtual_id_number":755932,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-05 14:53:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2020-10-04 15:30:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-10-05 13:35:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-10-05 13:35:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-04 15:30:49.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-10-05 13:35:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-10-05 14:53:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c780cfece9978844186ece1cba42cf31714bf06e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-03 14:05:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce11a3fafcb3336e3bcdb953ffca626c7bbae10d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-03 14:10:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fb2aeafcf2d12dcf49ac12e96dbe6cd285e70fb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 15:27:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98a59490618a42d4907a47cfeae507d6 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/507a84c23ed841698138e94bd20aa517 : SUCCESS in 3m 29s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/91ae0111306c41ed86f4fac8e714440b : SUCCESS in 12m 31s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d921d138bc4d4861adfcbb22e7f93eb5 : SUCCESS in 7m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/265c69e08569404c80e0f3dfeda0447d : SUCCESS in 11m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3b0e2b1b4c964d14b1aa0c998a4ef9d5 : SUCCESS in 47m 38s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e1440bc378c84c83b9d6c4e9c2e49f68 : SUCCESS in 1h 13m 33s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/31b4473f9147464bb324b6c481b74623 : SUCCESS in 16m 18s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1580b028e57240f18729f593fd423a26 : SUCCESS in 15m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95e2ae119b4743e5a12b302888195629 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"6b4eeff63ac457295f4ed6d5eaf8e59ed06b6fae","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-04 15:30:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"323bad3b59e8ec853e0a93478f6a61544a7af577","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-05 13:35:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4996313a0f3295dc9b24da84f183574c0f0a8e74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 13:38:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eecd79c874e7a03a76e8d87df20e4af4e4d992f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 14:53:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2dbd087fff54159bb46864d2e66ec51 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8524104e462848e1b04e134dcca31dc2 : SUCCESS in 4m 31s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/2654a025fab74334aefe5a2c031c7ac4 : SUCCESS in 9m 49s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5a7fca31e24439f9b1760b24788bca1 : SUCCESS in 9m 03s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2ad6e59031274846b69b902c4fe11473 : SUCCESS in 46m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/12971f7f6e094c208cc392e108addedd : SUCCESS in 1h 08m 00s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/bcc11e9c00024be0843c58563da7dd68 : SUCCESS in 20m 09s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06ecb223393a47b8afe853ac2f6eaff1 : SUCCESS in 14m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67f7280b1c0a494586b4b1c6c950d701 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"92022df96f479edddb8c3714ecec9a809b153065","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 14:53:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f3564ea8a516f7030bafd63752d16c21d2bf7c1f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 14:54:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dab53dc491ce4cddb5d95c5459c4ecf5 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"007d3a82f6bd72361066ea731a94a67c1b4d5f2b","revisions":{"f9555c400bb007e4a7d7da5c1965c645ce5987a2":{"kind":"REWORK","_number":1,"created":"2020-10-03 14:05:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/755932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/32/755932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/32/755932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/32/755932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/32/755932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/32/755932/1"}}},"commit":{"parents":[{"commit":"4620cd4dd10cda24c79a715bf0cd7eac48e6377c","subject":"Merge \"Fix rspec-puppet 2.7.1 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/4620cd4dd10cda24c79a715bf0cd7eac48e6377c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-03 14:02:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-03 14:04:37.000000000","tz":540},"subject":"Manage aodh-expirer package by aodh::expirer class","message":"Manage aodh-expirer package by aodh::expirer class\n\nSince aodh-expirer command is included in a different package, we need\nto manage that package as well in the corresponding class.\n\nChange-Id: Ie3111d5a168e8d15f3e04d410a5a5c30b44324c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f9555c400bb007e4a7d7da5c1965c645ce5987a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f9555c400bb007e4a7d7da5c1965c645ce5987a2"}]},"branch":"refs/heads/master"},"007d3a82f6bd72361066ea731a94a67c1b4d5f2b":{"kind":"REWORK","_number":2,"created":"2020-10-03 14:10:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/755932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/32/755932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/32/755932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/32/755932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/32/755932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/32/755932/2"}}},"commit":{"parents":[{"commit":"4620cd4dd10cda24c79a715bf0cd7eac48e6377c","subject":"Merge \"Fix rspec-puppet 2.7.1 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/4620cd4dd10cda24c79a715bf0cd7eac48e6377c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-03 14:02:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-03 14:10:15.000000000","tz":540},"subject":"Manage aodh-expirer package by aodh::expirer class","message":"Manage aodh-expirer package by aodh::expirer class\n\nSince aodh-expirer command is included in a different package, we need\nto manage that package as well in the corresponding class.\n\nChange-Id: Ie3111d5a168e8d15f3e04d410a5a5c30b44324c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/007d3a82f6bd72361066ea731a94a67c1b4d5f2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/007d3a82f6bd72361066ea731a94a67c1b4d5f2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
