)]}'
{"id":"openstack%2Fpuppet-aodh~758891","triplet_id":"openstack%2Fpuppet-aodh~master~I261c093be1c685dd57ac0eed1374d702cdb8c11b","project":"openstack/puppet-aodh","branch":"master","topic":"init-db","hashtags":[],"change_id":"I261c093be1c685dd57ac0eed1374d702cdb8c11b","subject":"Deprecate database options from the aodh class","status":"MERGED","created":"2020-10-20 01:22:32.000000000","updated":"2020-10-21 15:22:23.000000000","submitted":"2020-10-21 15:20:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758891-1603293613920-26dea723","meta_rev_id":"125be1daecac47cc0c94a61d624763bffbc43982","_number":758891,"virtual_id_number":758891,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"date":"2020-10-21 11:54:04.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-21 15:20:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-10-21 05:26:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-10-21 07:42:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-10-21 07:42:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-21 05:26:43.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-10-21 07:42:56.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-10-21 15:20:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"74cb2033f77683449915e53fad51cad915876c0f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-20 01:22:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"803b0567aa6f488addd1af138a7526bea1c45c48","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-20 02:01:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"718528d332e55337fbbe8bf375d1ce5b38a941a1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-20 02:04:18.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"051db90f885eb71ff0bfb37d4c305029a4c8015f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-21 00:46:29.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"06f4a0c05de952f06a4463f1311320c95c5c5ece","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 02:25:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1eccf095f2aa4e32aeaddf49d70f1463 : SUCCESS in 4m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e78f13d2f124e4593b6f21e0f9c9a46 : SUCCESS in 4m 20s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d1a83432c60b47688fc5d7dd8a388810 : SUCCESS in 10m 42s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf4ba0e460a2425ca7858df58f3afa56 : SUCCESS in 9m 09s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45b40d36b76b40f8a6cc136774224a65 : SUCCESS in 9m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ad70dd95d9844ec9a89a4617d8d8dca8 : SUCCESS in 52m 47s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ed78f9ba2fcd41c2bda5382efbc88682 : SUCCESS in 1h 29m 53s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/f8761bb3a0f449e0a61aff7a26b1d7d5 : SUCCESS in 20m 28s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6fec61be29940abb59e6feae168e405 : SUCCESS in 15m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6617f6d7dcda40c286fef02ae7b4f73f : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":3},{"id":"58bcafbf6787992417b1a776436f5fe607c01171","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-21 05:26:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"46079de8a4825c01ba352088bc0f9744a080e36e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-21 07:42:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"718b08b2f9297f0baa0defc61414c44e3e2d47e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 07:43:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"db4aad676c6efd1504038c975776df63328fecb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 09:10:20.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aba60f38a03c4a0096facc68d3a57ef1 : SUCCESS in 6m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a4110b8709a4aefa0939c51c700d9e3 : SUCCESS in 6m 37s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0935c6596ade41e381059b9cca47d911 : SUCCESS in 12m 26s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75cd240e108b49b8a2483b3154272275 : SUCCESS in 12m 57s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7aa1419c4e8f4a8182530093c09e23f7 : SUCCESS in 57m 13s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/317548f0f51846f2a443898d81cad949 : FAILURE in 1h 24m 47s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/91714bdf3ce745098d688940c2cd4f65 : SUCCESS in 22m 12s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45f71bc330484c59859ab15709b10f2a : SUCCESS in 17m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60e8003ff5ad4bd7ad0e2af23f80d7af : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":3},{"id":"6fcf6329a5e3b7932c8fc5606c9ae23aeba9b854","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-21 09:28:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9862f529b193397489d9da1dc478aee29b923ebf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 11:38:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/901d10a344af4bfd9446feb3e6ae7537 : SUCCESS in 5m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c02e0b67129b474fb14fd730f91c8b02 : SUCCESS in 4m 44s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0c79b8ae473047aba5f8685705bcb9a7 : SUCCESS in 12m 22s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1fd9a1af7f454733bbb2203eb7d10a86 : SUCCESS in 9m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e1556a9244447509b9c3ca62013db48 : SUCCESS in 11m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9b3e2ebd5be34d519845acdbce3bdd26 : FAILURE in 33m 47s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5ea6f250b2824968803ec0ebdc02271d : SUCCESS in 1h 19m 42s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d67da2c607b84826a05082d2f26f2979 : RETRY_LIMIT in 20m 31s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a21677973b904d7e92c68ee2b7073d98 : SUCCESS in 14m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1030747318e46939006021557b2c398 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":3},{"id":"d91488179c9e3d7f59aa422d7793579f1d89bfec","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-21 11:54:04.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7512c76fdb3291eeee3f96841a58dd3faf6c0297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 13:29:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9386d3a59d645b094d3d0e36da79b13 : SUCCESS in 4m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dfce8cf59d504aff8fbc5cbe38f772cf : SUCCESS in 5m 07s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/89a3f624a8fd49eba7c0cde0c2be1e39 : SUCCESS in 10m 48s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5f54ca8aa5d484597c28a3cd4347885 : SUCCESS in 8m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d117dbd3d63f4763bb0b98ccdefb95da : SUCCESS in 9m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/44c9d46d974241209705a9232db38174 : SUCCESS in 54m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/11ffc0bc42a8443cb92affe23999fc78 : SUCCESS in 1h 13m 57s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/927a435949cf47bba25d36dcd0b4aa3e : SUCCESS in 22m 50s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c0d92659a8c4bd9b1e737b4f81ac8c7 : SUCCESS in 15m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96bc838c6ac34e82897714bc72cf566f : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":3},{"id":"1d8a00ed91bf9737be5cf6afe721766e0842f254","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 13:29:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e56cb12d99588c74d34decd7c08eb62a141b20e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 15:20:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9348d1cbdb8455898eff608d5368890 : SUCCESS in 6m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d852162d13eb4c33b9499118135e0720 : SUCCESS in 6m 27s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d43676478c6340fea730c83b1b6cc097 : SUCCESS in 12m 28s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd5ac10b0d604748927023c7a7fbf40f : SUCCESS in 10m 33s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/22d76b8ae27e44e2b079d59a0e388322 : SUCCESS in 52m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/30af1dd794894b649e805920584a6e14 : SUCCESS in 1h 42m 46s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/6544623481384eb288b905a14d7cbcaf : SUCCESS in 21m 13s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd2ecdd9fa934c0bbf64d2b15de9340d : SUCCESS in 16m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adc0a0059a9c439899ee967c69bf1250 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":3},{"id":"756ce5e09364b9ea92568c4f1facf0d99ad6fc07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 15:20:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e944fdfaf67c0f4d98e7af02d5e2002ac2b91b32","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 15:22:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d77b2963431f4c57bb2fbeb87c9be302 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f9b9656b1744e0554770d94ac70cbbe7435639e1","revisions":{"e2e28550631bc01414df016fab33c066105323ae":{"kind":"REWORK","_number":1,"created":"2020-10-20 01:22:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/758891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/91/758891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/758891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/758891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/758891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/91/758891/1"}}},"commit":{"parents":[{"commit":"f40c746c90f96af6d91f58235086efc9c29c5c86","subject":"Bump version for the start of Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f40c746c90f96af6d91f58235086efc9c29c5c86"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-20 01:21:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-20 01:21:49.000000000","tz":540},"subject":"Deprecate database options from the aodh class","message":"Deprecate database options from the aodh class\n\n... because these options have been implemented in the aodh::db class.\n\nChange-Id: I261c093be1c685dd57ac0eed1374d702cdb8c11b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e2e28550631bc01414df016fab33c066105323ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e2e28550631bc01414df016fab33c066105323ae"}]},"branch":"refs/heads/master"},"c3c0db2b4082269deb6206a8eb9acbbf6cb074fe":{"kind":"REWORK","_number":2,"created":"2020-10-20 02:01:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/758891/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/91/758891/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/758891/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/758891/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/758891/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/91/758891/2"}}},"commit":{"parents":[{"commit":"f40c746c90f96af6d91f58235086efc9c29c5c86","subject":"Bump version for the start of Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f40c746c90f96af6d91f58235086efc9c29c5c86"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-20 01:21:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-20 02:01:07.000000000","tz":0},"subject":"Deprecate database options from the aodh class","message":"Deprecate database options from the aodh class\n\n... because these options have been implemented in the aodh::db class.\n\nChange-Id: I261c093be1c685dd57ac0eed1374d702cdb8c11b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c3c0db2b4082269deb6206a8eb9acbbf6cb074fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c3c0db2b4082269deb6206a8eb9acbbf6cb074fe"}]},"branch":"refs/heads/master"},"f9b9656b1744e0554770d94ac70cbbe7435639e1":{"kind":"REWORK","_number":3,"created":"2020-10-20 02:04:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/758891/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/91/758891/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/758891/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/758891/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/758891/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/91/758891/3"}}},"commit":{"parents":[{"commit":"f40c746c90f96af6d91f58235086efc9c29c5c86","subject":"Bump version for the start of Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f40c746c90f96af6d91f58235086efc9c29c5c86"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-20 01:21:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-20 02:03:54.000000000","tz":0},"subject":"Deprecate database options from the aodh class","message":"Deprecate database options from the aodh class\n\n... because these options have been implemented in the aodh::db class.\n\nChange-Id: I261c093be1c685dd57ac0eed1374d702cdb8c11b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f9b9656b1744e0554770d94ac70cbbe7435639e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f9b9656b1744e0554770d94ac70cbbe7435639e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
