)]}'
{"id":"openstack%2Fpuppet-aodh~759894","triplet_id":"openstack%2Fpuppet-aodh~master~I576b029fa3b0e765a9affd175f4113941d59abc7","project":"openstack/puppet-aodh","branch":"master","topic":"drop-standalone-api","hashtags":[],"change_id":"I576b029fa3b0e765a9affd175f4113941d59abc7","subject":"Drop support for standalone api service in CentOS/Ubuntu","status":"MERGED","created":"2020-10-27 16:01:15.000000000","updated":"2021-04-27 11:06:45.000000000","submitted":"2021-04-27 11:05:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759894-drop-standalone-api","meta_rev_id":"8c974c0c63ba31b042500a70be2b92990228450e","_number":759894,"virtual_id_number":759894,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-27 11:04:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2020-10-29 13:33:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-04-27 09:07:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-04-27 09:07:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 15:37:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-29 13:33:16.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-04-27 09:07:33.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"970cb8d997100ebbee3af24a907a56776ceb4049","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-27 16:01:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"327ceaa66ed1e9c930502f85f85b6d92a304b21e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-27 16:03:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8ab4ad4c9bfb35040f9827b5f4f3b6959fde8b78","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-27 16:04:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"62343173937404abc76204e43bcde679c4fb5fec","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-27 16:05:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"01728859dd65caf9d6d57b9c75310f829e1aa708","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-27 16:06:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5a7e51866d95e4b9e1d5ddc19d356dcaae1bef45","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-27 16:30:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4bf5277506f3431408b94c45307ca43fda20e5a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 17:49:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58be307541dc4e558ea3df7ec5f5b3b2 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1f1e6fcd70444128b559f973002831e : SUCCESS in 3m 20s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d9f5d970e61648bdacba9d42834415d4 : FAILURE in 10m 41s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/424a43a4f48e4060a78afbe514191846 : FAILURE in 8m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a83a9d75e80e4154a62e98c6a9ed1df7 : FAILURE in 10m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/73dd5f15a0a44bdaaea0225ab774f92e : FAILURE in 6m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/369e7b724d60411486d9a46ff178f400 : SUCCESS in 1h 14m 44s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/aab2796782924dae9813fc7c4c5ad8ec : SUCCESS in 20m 37s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fab49e78dbe4abab5052a85d616bea6 : SUCCESS in 12m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d5976ed6e33452a8b725789ce8e1f44 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":6},{"id":"f3ab4712b160198560c98a785506931fdecc5c7a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-28 13:50:41.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9f67334beb1fc42a3a6c6847a16487c2c3f1e06c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 15:37:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9a5f4c331964b3a8dc49471bf859bb6 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1b7ba566a1b47b0877985624b9895fe : SUCCESS in 3m 56s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/43a06871e2174ad7b3f633f0aa64e094 : SUCCESS in 19m 39s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38360d13f6d64abbbaeb298f958c61f6 : SUCCESS in 7m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/472bd55c34c14df3909bef315570f30e : SUCCESS in 8m 37s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/df2abd45da7242c5b86a810ab30aead4 : FAILURE in 5m 56s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5cc3a6ba82754a0e88423d5dfdad2490 : SUCCESS in 1h 30m 40s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/87c1001462ac45ce866e00f356dc94f0 : SUCCESS in 23m 06s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb9b29e20bb34e0eb2698295cd860433 : SUCCESS in 18m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5374db3d7dcf4cfda89cbd7706611fa1 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":7},{"id":"bd2b81ec5644bcdcf2a6ee0f4f80d1376addf24b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-29 13:33:16.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"1418441268d0b0d2a2a9de01238f145d6e9e439c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-04-27 09:07:33.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"bdfe9601ead55461fe4cd5d9ff778e18ccfff983","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 09:08:09.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"b34e9dc0a5116aade0b61f801c211d76f5b3d118","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 11:04:49.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd17c9a99b7a4f639ea5c0019704dc9e : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbcd221214ce4438bfcc5f2f2ccca0fd : SUCCESS in 4m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e563e3679df4614af31cb394540d066 : SUCCESS in 11m 50s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f46ce5248fd47b493072264ec5a393f : SUCCESS in 8m 50s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c172c546bed49d784d1d33e2b7cc08e : SUCCESS in 1h 23m 44s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/c02a9be93a954b13b3bfdb347d321ff1 : SUCCESS in 21m 13s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36682f9f850c4c52bb5916c1072a47ae : SUCCESS in 12m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98157e00ff8b4fb49cb415e6a8497a85 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":7},{"id":"abd6e38bb588e6141d9ec18d28e0c4c7ca49c345","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 11:05:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"8c974c0c63ba31b042500a70be2b92990228450e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 11:06:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcbecf2fdcdf47f6a4441578359427a1 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"7ce34f01b62ad546f415bda19b4778a467ea8924","revisions":{"b3726d8331ece8806cffcfc76ba161eda8389515":{"kind":"REWORK","_number":1,"created":"2020-10-27 16:01:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/759894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/94/759894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/1"}}},"commit":{"parents":[{"commit":"f9b9656b1744e0554770d94ac70cbbe7435639e1","subject":"Deprecate database options from the aodh class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f9b9656b1744e0554770d94ac70cbbe7435639e1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-27 16:00:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-27 16:00:48.000000000","tz":540},"subject":"Drop support for standalone api service in CentOS/Ubuntu","message":"Drop support for standalone api service in CentOS/Ubuntu\n\nAodh no longer provides standalone api service and we should use\nhttpd+wsgi to server api in CentOS and Ubuntu.\nThis patch removes support for stnadalone api service in CentOS and\nUbuntu and deprecates several parameters only used in standalone\nsetup.\n\nNote that standalone service is still used in Debian since it provides\nservices based on uwsgi, so the support is not dropped in Debian.\n\nChange-Id: I576b029fa3b0e765a9affd175f4113941d59abc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b3726d8331ece8806cffcfc76ba161eda8389515"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b3726d8331ece8806cffcfc76ba161eda8389515"}]},"branch":"refs/heads/master"},"ce56214d626523f2d885674f0d711c58b54afebb":{"kind":"REWORK","_number":2,"created":"2020-10-27 16:03:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/759894/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/94/759894/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/2"}}},"commit":{"parents":[{"commit":"f9b9656b1744e0554770d94ac70cbbe7435639e1","subject":"Deprecate database options from the aodh class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f9b9656b1744e0554770d94ac70cbbe7435639e1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-27 16:00:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-27 16:03:17.000000000","tz":540},"subject":"Drop support for standalone api service in CentOS/Ubuntu","message":"Drop support for standalone api service in CentOS/Ubuntu\n\nAodh no longer provides standalone api service and we should use\nhttpd+wsgi to server api in CentOS and Ubuntu.\nThis patch removes support for stnadalone api service in CentOS and\nUbuntu and deprecates several parameters only used in standalone\nsetup.\n\nNote that standalone service is still used in Debian since it provides\nservices based on uwsgi, so the support is not dropped in Debian.\n\nChange-Id: I576b029fa3b0e765a9affd175f4113941d59abc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ce56214d626523f2d885674f0d711c58b54afebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ce56214d626523f2d885674f0d711c58b54afebb"}]},"branch":"refs/heads/master"},"e36da7346ba418e9c410d8e1694aedb8ded33b29":{"kind":"REWORK","_number":3,"created":"2020-10-27 16:04:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/759894/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/94/759894/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/3"}}},"commit":{"parents":[{"commit":"f9b9656b1744e0554770d94ac70cbbe7435639e1","subject":"Deprecate database options from the aodh class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f9b9656b1744e0554770d94ac70cbbe7435639e1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-27 16:00:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-27 16:04:16.000000000","tz":540},"subject":"Drop support for standalone api service in CentOS/Ubuntu","message":"Drop support for standalone api service in CentOS/Ubuntu\n\nAodh no longer provides standalone api service and we should use\nhttpd+wsgi to server api in CentOS and Ubuntu.\nThis patch removes support for stnadalone api service in CentOS and\nUbuntu and deprecates several parameters only used in standalone\nsetup.\n\nNote that standalone service is still used in Debian since it provides\nservices based on uwsgi, so the support is not dropped in Debian.\n\nChange-Id: I576b029fa3b0e765a9affd175f4113941d59abc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e36da7346ba418e9c410d8e1694aedb8ded33b29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e36da7346ba418e9c410d8e1694aedb8ded33b29"}]},"branch":"refs/heads/master"},"a0e30802dad6fb8f37c198ba659bd2465f46867e":{"kind":"REWORK","_number":4,"created":"2020-10-27 16:05:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/759894/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/94/759894/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/4"}}},"commit":{"parents":[{"commit":"f9b9656b1744e0554770d94ac70cbbe7435639e1","subject":"Deprecate database options from the aodh class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f9b9656b1744e0554770d94ac70cbbe7435639e1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-27 16:00:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-27 16:05:00.000000000","tz":540},"subject":"Drop support for standalone api service in CentOS/Ubuntu","message":"Drop support for standalone api service in CentOS/Ubuntu\n\nAodh no longer provides standalone api service and we should use\nhttpd+wsgi to server api in CentOS and Ubuntu.\nThis patch removes support for stnadalone api service in CentOS and\nUbuntu and deprecates several parameters only used in standalone\nsetup.\n\nNote that standalone service is still used in Debian since it provides\nservices based on uwsgi, so the support is not dropped in Debian.\n\nChange-Id: I576b029fa3b0e765a9affd175f4113941d59abc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/a0e30802dad6fb8f37c198ba659bd2465f46867e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/a0e30802dad6fb8f37c198ba659bd2465f46867e"}]},"branch":"refs/heads/master"},"b0b955c1a431912299beb1b3bded4a8b10345488":{"kind":"REWORK","_number":5,"created":"2020-10-27 16:06:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/759894/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/94/759894/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/5"}}},"commit":{"parents":[{"commit":"f9b9656b1744e0554770d94ac70cbbe7435639e1","subject":"Deprecate database options from the aodh class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f9b9656b1744e0554770d94ac70cbbe7435639e1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-27 16:00:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-27 16:06:52.000000000","tz":540},"subject":"Drop support for standalone api service in CentOS/Ubuntu","message":"Drop support for standalone api service in CentOS/Ubuntu\n\nAodh no longer provides standalone api service and we should use\nhttpd+wsgi to server api in CentOS and Ubuntu.\nThis patch removes support for stnadalone api service in CentOS and\nUbuntu and deprecates several parameters only used in standalone\nsetup.\n\nNote that standalone service is still used in Debian since it provides\nservices based on uwsgi, so the support is not dropped in Debian.\n\nChange-Id: I576b029fa3b0e765a9affd175f4113941d59abc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b0b955c1a431912299beb1b3bded4a8b10345488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b0b955c1a431912299beb1b3bded4a8b10345488"}]},"branch":"refs/heads/master"},"6c5eaf0f55c1890baad14653d15f6b909aea385b":{"kind":"REWORK","_number":6,"created":"2020-10-27 16:30:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/759894/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/94/759894/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/6"}}},"commit":{"parents":[{"commit":"f9b9656b1744e0554770d94ac70cbbe7435639e1","subject":"Deprecate database options from the aodh class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f9b9656b1744e0554770d94ac70cbbe7435639e1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-27 16:00:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-27 16:29:59.000000000","tz":540},"subject":"Drop support for standalone api service in CentOS/Ubuntu","message":"Drop support for standalone api service in CentOS/Ubuntu\n\nAodh no longer provides standalone api service and we should use\nhttpd+wsgi to server api in CentOS and Ubuntu.\nThis patch removes support for stnadalone api service in CentOS and\nUbuntu and deprecates several parameters only used in standalone\nsetup.\n\nNote that standalone service is still used in Debian since it provides\nservices based on uwsgi, so the support is not dropped in Debian.\n\nChange-Id: I576b029fa3b0e765a9affd175f4113941d59abc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6c5eaf0f55c1890baad14653d15f6b909aea385b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6c5eaf0f55c1890baad14653d15f6b909aea385b"}]},"branch":"refs/heads/master"},"7ce34f01b62ad546f415bda19b4778a467ea8924":{"kind":"REWORK","_number":7,"created":"2020-10-28 13:50:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/759894/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/94/759894/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/94/759894/7"}}},"commit":{"parents":[{"commit":"f9b9656b1744e0554770d94ac70cbbe7435639e1","subject":"Deprecate database options from the aodh class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f9b9656b1744e0554770d94ac70cbbe7435639e1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-27 16:00:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-28 13:50:33.000000000","tz":540},"subject":"Drop support for standalone api service in CentOS/Ubuntu","message":"Drop support for standalone api service in CentOS/Ubuntu\n\nAodh no longer provides standalone api service and we should use\nhttpd+wsgi to server api in CentOS and Ubuntu.\nThis patch removes support for stnadalone api service in CentOS and\nUbuntu and deprecates several parameters only used in standalone\nsetup.\n\nNote that standalone service is still used in Debian since it provides\nservices based on uwsgi, so the support is not dropped in Debian.\n\nChange-Id: I576b029fa3b0e765a9affd175f4113941d59abc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/7ce34f01b62ad546f415bda19b4778a467ea8924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/7ce34f01b62ad546f415bda19b4778a467ea8924"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
