)]}'
{"id":"openstack%2Fpuppet-aodh~765107","triplet_id":"openstack%2Fpuppet-aodh~master~I88e36bff5b24d81e687662e655fb6c5470a15ed6","project":"openstack/puppet-aodh","branch":"master","topic":"memcache-authtoken","hashtags":[],"change_id":"I88e36bff5b24d81e687662e655fb6c5470a15ed6","subject":"[WiP] Add TLS options to keystone authtoken/oslo.cache","status":"ABANDONED","created":"2020-12-02 10:53:24.000000000","updated":"2022-07-23 13:51:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fff2de28b17d25792c0facb631da74e72f2e7626","_number":765107,"virtual_id_number":765107,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-02 11:13:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":-1,"date":"2020-12-21 08:46:49.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-02 11:13:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c8179200990f4fa4bbe547446fc2ce2a3bbabc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2020-12-02 10:53:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b70f389e1fb070c41a0aee74840d08046f1282b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 11:13:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63c4614e59704c568bf49c31dab81e17 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd48a32799234f1e894510bb656b25d6 : SUCCESS in 3m 05s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/ddd14a278d0c4ba8bb273d7d764be80d : FAILURE in 13m 13s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b52a7affb99844ca8aa06a3def11592d : FAILURE in 7m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf995c9ab0ca43b18d7f9e35e5239868 : FAILURE in 10m 58s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa1d236c54cf4a5888464e3c9c7f4145 : FAILURE in 9m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/972bd4550d1e4e3eb05d51282b79a2c6 : FAILURE in 8m 57s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/f7f52051833e446cadb037f0a456750f : FAILURE in 8m 50s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72d0784c3d504a5ca0a6b9c7389719ab : FAILURE in 6m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"754c570f199102eeee4b5387a20b6b455050ba17","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2020-12-21 08:46:49.000000000","message":"Patch Set 1: Workflow-1\n\nw-1 until keystonemiddleware patch is merged https://review.opendev.org/c/openstack/keystonemiddleware/+/766735","accounts_in_message":[],"_revision_number":1},{"id":"81dec4f297cbb1e974ece7cc7a243dff9dbb2f95","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-02-04 19:06:21.000000000","message":"Topic memcached-cert removed","accounts_in_message":[],"_revision_number":1},{"id":"f1d9ee0bed1c5cf85f45487bbd5fa9308fd4e9fe","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-02-04 19:06:35.000000000","message":"Topic set to memcache-authtoken","accounts_in_message":[],"_revision_number":1},{"id":"fff2de28b17d25792c0facb631da74e72f2e7626","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-23 13:51:32.000000000","message":"Abandoned\n\nThis patch has been inactive for more than 1 year. The dependent patch has not been merged. So I\u0027ll abandon this to clean up our review queue.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ef8b1e25cc5729c3be5cb76f2861ec044a95ce9","revisions":{"8ef8b1e25cc5729c3be5cb76f2861ec044a95ce9":{"kind":"REWORK","_number":1,"created":"2020-12-02 10:53:24.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/07/765107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/07/765107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/07/765107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/07/765107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/07/765107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/07/765107/1"}}},"commit":{"parents":[{"commit":"75ff328e89c2e975f8eea12f6ed8d4e9a07f45a4","subject":"Allow db sync timeouts to be configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/75ff328e89c2e975f8eea12f6ed8d4e9a07f45a4"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2020-12-02 10:52:42.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2020-12-02 10:53:22.000000000","tz":60},"subject":"[WiP] Add TLS options to keystone authtoken/oslo.cache","message":"[WiP] Add TLS options to keystone authtoken/oslo.cache\n\nThis patch specifies a set of options required to build a TLS context.\nThe context built from those options can later on be passed to any of\nthe oslo.cache backends that supports TLS connections.\n\nChange-Id: I88e36bff5b24d81e687662e655fb6c5470a15ed6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8ef8b1e25cc5729c3be5cb76f2861ec044a95ce9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8ef8b1e25cc5729c3be5cb76f2861ec044a95ce9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
