)]}'
{"id":"openstack%2Fpuppet-aodh~768776","triplet_id":"openstack%2Fpuppet-aodh~master~Id64fc407d19aa546512078f67df3a727fd9f5525","project":"openstack/puppet-aodh","branch":"master","topic":"coordination_url-default","hashtags":[],"change_id":"Id64fc407d19aa546512078f67df3a727fd9f5525","subject":"Add support for workers option","status":"MERGED","created":"2020-12-30 11:50:12.000000000","updated":"2021-01-25 16:28:48.000000000","submitted":"2021-01-25 16:27:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768776-coordination_url-default","meta_rev_id":"abdf865593eda47af559d31bdd4158b325934c4b","_number":768776,"virtual_id_number":768776,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-25 16:27:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-01-25 14:56:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-01-25 08:40:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2021-01-25 14:56:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-30 13:14:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-04 14:33:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-01-05 16:00:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"60d2424c9426230f763afeeefb5cfa547d40c986","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-30 11:50:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04b277f27358483564fe401b19e9db0fb8b7e9a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-30 13:14:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/672ba373c74a4ca08ae6ceb785c5b192 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d53b68c4085e4b399c222cc232edc701 : SUCCESS in 2m 52s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/a70a234342074759864ed6e71561532e : FAILURE in 9m 18s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86b7a5681b3a4f07ac4d73e046f50eee : FAILURE in 7m 08s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b787f4e8203466e8d83506ba1be96db : FAILURE in 7m 46s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c149e0235514181885a8d3c2f8c262e : SUCCESS in 56m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6fb57f9f5eb1416bb3795cd4f9414e39 : SUCCESS in 1h 20m 37s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/f597b3a5412f49d98e4a0b90e04eb00a : SUCCESS in 17m 07s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5296658a051748a1b23d487ae9ba3710 : SUCCESS in 12m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b6ce2cbf6f74820a380ad3f7d7613f1 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6f4e217d4ef125a3028b55ddc48d9b65f4d4ff56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-31 01:50:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5d6ab3442b577611080385498b8c1ad9a180e012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-31 03:35:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df9007c20f6244aa8fd770a92ff9c42c : SUCCESS in 3m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cec756b3da6140ba8d70dc3ecf2ef3e3 : SUCCESS in 2m 50s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/219dc1a2de7e4b2b969200ca109b368b : SUCCESS in 8m 55s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44fbf766c13d4545bc4f5845b117dd20 : SUCCESS in 7m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2d1c43eb6ad49e79a2784165ae15f7b : SUCCESS in 7m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95ead79641944718b652e3928076124c : SUCCESS in 55m 13s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2b3c4a8a4bb3455aa52f4ccc57e94745 : SUCCESS in 1h 41m 23s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/795994d41de744959b633d09f00ae8cf : SUCCESS in 16m 26s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96df5c20e2a64b8b8be744be63343388 : SUCCESS in 11m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c2647dfd76242d49c724c12f1da7802 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":2},{"id":"7b947f46e748537f73cfd5c2325050efe0cacee1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-04 14:33:37.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa237f19966b66ddde27b69dbbec7ca6e7b059c0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-01-05 16:00:16.000000000","message":"Patch Set 2: Code-Review+2\n\nThis is fine, we can do as Alex suggested but probably best to only do on master if you want this backportable to not reload everybody\u0027s services if we change the default workers count :)","accounts_in_message":[],"_revision_number":2},{"id":"7665ea534c1bb38667411ef7c75b381d6a36dffd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-14 15:12:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"840cd99f9c0c83c12bab035acfe163f69150cd17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 15:29:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d6d0f6c0f4540f391677a777d8f8bd7 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c216dc1386294ca280921acf0d32843e : FAILURE in 3m 01s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d4e52f60e9574b418e661e8254b98789 : FAILURE in 9m 15s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2afe2c3e6344bc795a91a1966ae3b95 : FAILURE in 9m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/062e8101d2084cefa8071f6412053334 : FAILURE in 9m 55s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a20a5e8836504950aef977a97339ba91 : FAILURE in 9m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a4acfc00e1bb44058b0bdd9beb3f4ab0 : FAILURE in 9m 07s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/2b9f6459434c4974883ddd671be85c3f : FAILURE in 10m 10s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ccf2d732ea6246d79ec7e76e9d297203 : FAILURE in 6m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b519a54586b64179802306d0cffcf15e : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":3},{"id":"fb1da413d60c00f99d9c86fe41f9ff247be48b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-14 15:36:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5f5316e674116e1aa2012f0f3fb167a6b227b771","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-14 15:36:47.000000000","message":"Topic coordination_url-default removed","accounts_in_message":[],"_revision_number":4},{"id":"b054f9eeba23a40d00099752fe8f8cdbd046cd0f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-14 15:37:04.000000000","message":"Topic set to workers","accounts_in_message":[],"_revision_number":4},{"id":"3222f163c1e8d7558de13fc9c2f2aecdd2b863bd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-14 15:38:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c88428dd68369134644e7169bd8206c7dc3e1bb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:28:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e99a8361b2cf436b969c6d926343c93d : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c29e48796184bb1a0530dd79b7f04b8 : SUCCESS in 3m 41s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0b63ec2980cc4b06af1a3b6af71bd6a0 : FAILURE in 12m 09s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd93215181f54b72bdd9491b70d76fd1 : FAILURE in 9m 41s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9e154c58c524aafb23043147e1482d6 : FAILURE in 7m 55s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7120cf172e745c18837252c72a4021d : SUCCESS in 53m 22s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c799c34b0a884027834bd41609156161 : SUCCESS in 1h 31m 59s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/01fcc77a70a54dd98f083606accd3240 : SUCCESS in 17m 51s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6653d2b1f0824132a8514ed7379d32b6 : SUCCESS in 12m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af5cf1a360c74134ba29f2c29667ac12 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":4},{"id":"ce35d722786b6d723be8f39faa0bd67dbc56edea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-18 06:58:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b7847a702352c0d7d0834c5a8b642fe02873a9fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 07:50:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/824351baaa82416ea9dd666c47f369db : RETRY_LIMIT in 2m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca5e41dc7cae44fe88b373e26ff39f22 : RETRY_LIMIT in 1m 58s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9e9548a6bf1946a78f4f02813e2ea9e8 : SUCCESS in 8m 07s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3b2ee0ba2fb426a983a03fd2d00df4f : RETRY_LIMIT in 5m 11s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0aeca1cc72114542a922858e25ae45db : RETRY_LIMIT in 6m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49656e47224b42d6920bbaabff854d38 : RETRY_LIMIT in 4m 20s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0e6feccd76e1461f891bf27acbd1739d : RETRY_LIMIT in 5m 03s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/55b4e3cc4aea400297c291bcf7a92d0c : RETRY_LIMIT in 3m 36s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b124795ebd3e4ebd8cb37bc12b665d6b : RETRY_LIMIT in 4m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/719bf3b1d819414a8152b3c1c265f557 : RETRY_LIMIT in 1m 33s","accounts_in_message":[],"_revision_number":5},{"id":"569af89fa4a247175a9804ee76627a527ef20674","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-23 03:36:11.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"95eee3ba4296e3e2eca8b8e9f06556128bff2c3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-23 04:36:05.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1d2c8efdb4c410a9828ce4a1b2e1e4d : SUCCESS in 3m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f58031ca9ad34d5290219f377b981c5e : SUCCESS in 2m 58s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d536ef73410c4453b4ea42da40714b82 : SUCCESS in 11m 31s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4db5e24bf86945b699518d500b183441 : SUCCESS in 7m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/beaadac8d4ff4867a2a429e0526bf8c4 : SUCCESS in 9m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e93b27d20494517b170c54406253245 : SUCCESS in 57m 01s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6a3ce34dd1054dc0a60ef8c8cbf14237 : FAILURE in 10m 16s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/00789451a80447b8a0e269deb1697d0d : SUCCESS in 17m 38s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3dbb4c228234dc3a08f2f01c56c1cb8 : SUCCESS in 15m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0087ebee9894183af8a169bc10b82fb : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":5},{"id":"d70598ae520f294a226d51876e1a0e246984a0f9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-25 03:46:50.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ac118b0e73d985b4e06275ae1b0397f038ba5817","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 05:39:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/546b9f33164e49acad8fb8092766d6af : SUCCESS in 3m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7301bb74b9cb4310bf96086aada86a53 : SUCCESS in 3m 12s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/8a09301a31d24420a091a7065e4da453 : SUCCESS in 8m 58s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a51c1789d0d44538f178a90796b5f2a : SUCCESS in 5m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a595e7823fa942d48b75c7fc84434b2c : SUCCESS in 7m 29s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9adf6977eb5d4c7ab01fd13f9ba25712 : SUCCESS in 56m 33s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6b8353e5ff0349808475d0d011b31e8c : SUCCESS in 1h 26m 28s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/ae90cd62999c4d06a8bebf9c5620ef60 : SUCCESS in 15m 00s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a0e0be619d8466a8529f0a86b528bed : SUCCESS in 13m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1ad5bcd70c64a639db3409f4aca9122 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":5},{"id":"ad3ec9026d9d64c545951f941d1bde01728bf86f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-01-25 08:40:59.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"6bd85ee10a3a406dcd1df73ed2f525a206178dbb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-25 14:56:48.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c5eabce2f5c1d742dbeefeea5cdb7cf651b54bc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 14:57:11.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f7a0eaa841b437c824387ee6f17c27a49546d22a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 16:27:17.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee760255fd1d4d5e9947e181d64ccadb : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbd5e27da7534590a0ee30a1492b6f98 : SUCCESS in 3m 33s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1b10cd1339764c838e37446e70915778 : SUCCESS in 8m 47s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ddddf142c5b54f63953a55957e543216 : SUCCESS in 6m 21s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a540a8a61d40437b9bd922d9daa0b927 : SUCCESS in 1h 00m 20s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/118a4dc1f2b640098a6fef0059bb27d2 : SUCCESS in 1h 23m 27s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/14fd18c649ff4b8ca4e33388d7c38419 : SUCCESS in 19m 40s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e96378dc4fc74cfe84d87e24e14d697f : SUCCESS in 12m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdbc9f6ce7714c95b0d40648a2f63c2c : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":5},{"id":"f2b274f14dbcf7ec9545f2830423c3f3b884cd7f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 16:27:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"abdf865593eda47af559d31bdd4158b325934c4b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 16:28:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fdf32a3b46f47ffa46cbf7c7435844f : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ae9b91107dffa80c3674a80f6c3a28ffa403d5bb","revisions":{"aac6c89c16e0ed14b77a3b19561708fbead6fc4f":{"kind":"REWORK","_number":1,"created":"2020-12-30 11:50:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/768776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/76/768776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/1"}}},"commit":{"parents":[{"commit":"99eb1885129b06183beefa48f7ebdd70ee33ff72","subject":"Prepare Wallaby M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/99eb1885129b06183beefa48f7ebdd70ee33ff72"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-30 11:48:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-30 11:50:04.000000000","tz":540},"subject":"Add support for workers option","message":"Add support for workers option\n\nThis change introduces support for workers option of aodh services\n(evaluator, listener, notifier) so that operators can define number\nof processes used in each service.\n\nChange-Id: Id64fc407d19aa546512078f67df3a727fd9f5525\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/aac6c89c16e0ed14b77a3b19561708fbead6fc4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/aac6c89c16e0ed14b77a3b19561708fbead6fc4f"}]},"branch":"refs/heads/master"},"e1cb3933a1deff5980baa1afe2ccbd8a314ad030":{"kind":"REWORK","_number":2,"created":"2020-12-31 01:50:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/768776/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/76/768776/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/2"}}},"commit":{"parents":[{"commit":"99eb1885129b06183beefa48f7ebdd70ee33ff72","subject":"Prepare Wallaby M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/99eb1885129b06183beefa48f7ebdd70ee33ff72"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-30 11:48:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-31 01:49:51.000000000","tz":540},"subject":"Add support for workers option","message":"Add support for workers option\n\nThis change introduces support for workers option of aodh services\n(evaluator, listener, notifier) so that operators can define number\nof processes used in each service.\n\nChange-Id: Id64fc407d19aa546512078f67df3a727fd9f5525\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e1cb3933a1deff5980baa1afe2ccbd8a314ad030"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e1cb3933a1deff5980baa1afe2ccbd8a314ad030"}]},"branch":"refs/heads/master"},"18caaeaf54477e6255eb1398a974ab1ee3baeb7f":{"kind":"REWORK","_number":3,"created":"2021-01-14 15:12:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/768776/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/76/768776/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/3"}}},"commit":{"parents":[{"commit":"99eb1885129b06183beefa48f7ebdd70ee33ff72","subject":"Prepare Wallaby M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/99eb1885129b06183beefa48f7ebdd70ee33ff72"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-30 11:48:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-14 15:12:03.000000000","tz":540},"subject":"Add support for workers option","message":"Add support for workers option\n\nThis change introduces support for workers option of aodh services\n(evaluator, listener, notifier) so that operators can define number\nof processes used in each service.\n\nChange-Id: Id64fc407d19aa546512078f67df3a727fd9f5525\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/18caaeaf54477e6255eb1398a974ab1ee3baeb7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/18caaeaf54477e6255eb1398a974ab1ee3baeb7f"}]},"branch":"refs/heads/master"},"33773c16c1356a2d95f98fa471ca4908ade663db":{"kind":"REWORK","_number":4,"created":"2021-01-14 15:36:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/768776/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/76/768776/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/4"}}},"commit":{"parents":[{"commit":"99eb1885129b06183beefa48f7ebdd70ee33ff72","subject":"Prepare Wallaby M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/99eb1885129b06183beefa48f7ebdd70ee33ff72"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-30 11:48:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-14 15:33:24.000000000","tz":540},"subject":"Add support for workers option","message":"Add support for workers option\n\nThis change introduces support for workers option of aodh services\n(evaluator, listener, notifier) so that operators can define number\nof processes used in each service.\n\nChange-Id: Id64fc407d19aa546512078f67df3a727fd9f5525\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/33773c16c1356a2d95f98fa471ca4908ade663db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/33773c16c1356a2d95f98fa471ca4908ade663db"}]},"branch":"refs/heads/master"},"ae9b91107dffa80c3674a80f6c3a28ffa403d5bb":{"kind":"REWORK","_number":5,"created":"2021-01-18 06:58:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/768776/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/76/768776/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/76/768776/5"}}},"commit":{"parents":[{"commit":"99eb1885129b06183beefa48f7ebdd70ee33ff72","subject":"Prepare Wallaby M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/99eb1885129b06183beefa48f7ebdd70ee33ff72"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-30 11:48:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-18 03:57:33.000000000","tz":540},"subject":"Add support for workers option","message":"Add support for workers option\n\nThis change introduces support for workers option of aodh services\n(evaluator, listener, notifier) so that operators can define number\nof processes used in each service.\n\nChange-Id: Id64fc407d19aa546512078f67df3a727fd9f5525\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ae9b91107dffa80c3674a80f6c3a28ffa403d5bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ae9b91107dffa80c3674a80f6c3a28ffa403d5bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
