)]}'
{"id":"openstack%2Fpuppet-aodh~772689","triplet_id":"openstack%2Fpuppet-aodh~master~I876d02605d346f4680672f9f4832bfbed91c4235","project":"openstack/puppet-aodh","branch":"master","topic":"service_credentials","hashtags":[],"change_id":"I876d02605d346f4680672f9f4832bfbed91c4235","subject":"Remove the deprecated parameters of aodh::auth","status":"MERGED","created":"2021-01-27 13:06:00.000000000","updated":"2021-02-01 10:04:34.000000000","submitted":"2021-02-01 10:02:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772689-service_credentials","meta_rev_id":"a05c74cfe653ba51a52599ef83ab8c671d3447ea","_number":772689,"virtual_id_number":772689,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-01 10:02:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-02-01 08:35:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-02-01 10:04:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2021-02-01 08:35:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-27 14:43:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-27 21:32:05.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-02-01 08:35:59.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"d71ce5151146a913f1c706a7b42e1fbddbcf8d30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-27 13:06:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05a87871eae2ef90efb465d6a1c62b05940cfec5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 14:43:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ee8fa13cda841f69626c8fb65c21052 : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9cc1e6cba32d47689e62ebd85a3efdeb : SUCCESS in 3m 16s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9a08ac578cb547dd93e74204aef447e8 : SUCCESS in 9m 17s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5d5792505b344e3828efe6ef90b8391 : SUCCESS in 6m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ea76eebf1ae461b87c9e0cf870f598d : SUCCESS in 10m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8fc7eb6502642f7a07f290281a7eff8 : SUCCESS in 1h 02m 13s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/749bf91a15c443d78f602bdb044f3cbc : SUCCESS in 1h 30m 57s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/05b631b1ee934ba79111cd1b38d00d47 : SUCCESS in 18m 15s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1764492405fb44fa889aacd52b8d8794 : SUCCESS in 12m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ae168c6a74741c49cae5097ba0a508f : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"f0092e6276f66ef05008361c0f76e13a3f5d03e7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-27 21:32:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c0d7fd03f9aca927f595fbe8f81d06add72b4cd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-28 00:47:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"7e77b3e38ce09990bba99ca5ed8f119d03492b33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 02:16:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87c38de9e44347abb0efdcb8155a400f : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f51a367a1044b44bf156a650df0dbdb : SUCCESS in 3m 00s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/a62688ae401b4230b7de9c74afc29e59 : SUCCESS in 9m 25s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37bf8eaf530b412ca28d8c2a351f6a54 : SUCCESS in 6m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03f854985d53426ca150638dd732ebc2 : SUCCESS in 7m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/242e5c20400e46538ebaeceab8cc06b5 : SUCCESS in 53m 09s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c2dec0487272419da4c4190913a1ab6f : SUCCESS in 1h 23m 31s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/8827f51abe4d40ccaf793b2e45bcbce5 : SUCCESS in 16m 40s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d4e055de30c47a9be3b096787177051 : SUCCESS in 12m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7b3d013d7ea4fa78e8ec6b4c967eac8 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":2},{"id":"84af093d03dd7f34ee33133e6c34504a8ee0e2d1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-02-01 08:35:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2be24844e417394a3693a66107245da54f3c485d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 08:36:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ff1e01035f86f59d40120884f3bfc8dd66a18a93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 10:02:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4357da5b2e244d53b44d52394439e761 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9003fd9b1c884eb097a599cdb36bfed9 : SUCCESS in 2m 45s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c12f71a165d24b84ad1d934b91aa99ed : SUCCESS in 8m 54s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd71be228c41435b84ccab53a27e8b29 : SUCCESS in 6m 01s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b3352dad31f49249d986303c4de72b9 : SUCCESS in 1h 03m 59s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/14f6a399f6724b8489d9d2829b0f5b6d : SUCCESS in 1h 23m 40s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/6d7fcf7834c9471bb3fc6992d1e6e31d : SUCCESS in 19m 23s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39b5da70f798459fa76068f8bc8fabef : SUCCESS in 13m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e089ec83f2f14cd29215ccfd32f85245 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"3b1aa2b9b8f5f7d08c0fababdd116493752524af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 10:02:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"73aa62b8483bace4bcdfaa4095213d22c98d4333","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 10:04:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/192a7c6e5f604babad8881985a412d6a : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"77691c4c311c1e39418ba22de9b56f6bd43cf9fd","revisions":{"4124fdf993f93ea6c88975ce8a7d558524293278":{"kind":"REWORK","_number":1,"created":"2021-01-27 13:06:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/772689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/89/772689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/89/772689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/89/772689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/89/772689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/89/772689/1"}}},"commit":{"parents":[{"commit":"99eb1885129b06183beefa48f7ebdd70ee33ff72","subject":"Prepare Wallaby M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/99eb1885129b06183beefa48f7ebdd70ee33ff72"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-27 13:05:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-27 13:05:39.000000000","tz":540},"subject":"Remove the deprecated parameters of aodh::auth","message":"Remove the deprecated parameters of aodh::auth\n\nChange-Id: I876d02605d346f4680672f9f4832bfbed91c4235\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/4124fdf993f93ea6c88975ce8a7d558524293278"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/4124fdf993f93ea6c88975ce8a7d558524293278"}]},"branch":"refs/heads/master"},"77691c4c311c1e39418ba22de9b56f6bd43cf9fd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-28 00:47:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/772689/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/89/772689/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/89/772689/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/89/772689/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/89/772689/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/89/772689/2"}}},"commit":{"parents":[{"commit":"94555d4872d7033b6d5ca497f02c44237ea36d86","subject":"Merge \"Ensure coordination_url is unset by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/94555d4872d7033b6d5ca497f02c44237ea36d86"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-27 13:05:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-28 00:47:14.000000000","tz":540},"subject":"Remove the deprecated parameters of aodh::auth","message":"Remove the deprecated parameters of aodh::auth\n\nChange-Id: I876d02605d346f4680672f9f4832bfbed91c4235\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/77691c4c311c1e39418ba22de9b56f6bd43cf9fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/77691c4c311c1e39418ba22de9b56f6bd43cf9fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
