)]}'
{"id":"openstack%2Fpuppet-aodh~772840","triplet_id":"openstack%2Fpuppet-aodh~master~I62c02435ae2f8712327a3bf07de56f9e3d693548","project":"openstack/puppet-aodh","branch":"master","topic":"c8s","hashtags":[],"change_id":"I62c02435ae2f8712327a3bf07de56f9e3d693548","subject":"DNM Test c8s","status":"ABANDONED","created":"2021-01-28 08:58:41.000000000","updated":"2022-03-13 14:46:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3d5017d6447b0b5cd2b6d3fae7930592f2f6eef9","_number":772840,"virtual_id_number":772840,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-28 10:57:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-28 10:57:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1016da3224dc5bde215c9ba7864c39a4fbbeaf89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-01-28 08:58:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71228b39a51cec9af44209a3e19300a3629fe9ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 10:57:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad4dedbc28664fa698de5b5764a9ea64 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ef33eba585540fbbb99c9c8824fd7f6 : SUCCESS in 3m 36s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/810f1dc77ca94c939f054ee743e5ee36 : SUCCESS in 9m 50s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/f998ae9a188d4105abf2cf42584e9e03 : SUCCESS in 9m 24s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3601a4acccf74ebea83ca5622b5a2c86 : SUCCESS in 7m 29s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07f82a8bb18b40e2bbed118817482e2b : SUCCESS in 12m 29s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97709f852a3e4431b87b6a548ae9beaa : SUCCESS in 51m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9f4a6d834f61421aa1b73aa63acace1d : SUCCESS in 1h 36m 38s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/47dfcabc60884c01b4fd4d408d07d076 : SUCCESS in 1h 25m 17s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/0b03e2c7a9db4fc8b816f48b003998cc : SUCCESS in 15m 56s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c9a89e61c394b3da075a388a30b5e44 : SUCCESS in 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d5017d6447b0b5cd2b6d3fae7930592f2f6eef9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-13 14:46:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8271d43f302250a4772ea47b1826534864f8023a","revisions":{"8271d43f302250a4772ea47b1826534864f8023a":{"kind":"REWORK","_number":1,"created":"2021-01-28 08:58:41.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/40/772840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/40/772840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/772840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/772840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/772840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/40/772840/1"}}},"commit":{"parents":[{"commit":"94555d4872d7033b6d5ca497f02c44237ea36d86","subject":"Merge \"Ensure coordination_url is unset by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/94555d4872d7033b6d5ca497f02c44237ea36d86"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-01-28 08:58:12.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-01-28 08:58:39.000000000","tz":60},"subject":"DNM Test c8s","message":"DNM Test c8s\n\nDepends-On: https://review.opendev.org/c/openstack/puppet-openstack-integration/+/772839\nChange-Id: I62c02435ae2f8712327a3bf07de56f9e3d693548\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8271d43f302250a4772ea47b1826534864f8023a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8271d43f302250a4772ea47b1826534864f8023a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
