)]}'
{"id":"openstack%2Fpuppet-aodh~779885","triplet_id":"openstack%2Fpuppet-aodh~master~Ia63c6e752a75d8acbcc3a2fc3b34c6ace65fc129","project":"openstack/puppet-aodh","branch":"master","hashtags":[],"change_id":"Ia63c6e752a75d8acbcc3a2fc3b34c6ace65fc129","subject":"Fix the error raised when password is not set","status":"MERGED","created":"2021-03-11 01:08:00.000000000","updated":"2021-03-12 16:35:36.000000000","submitted":"2021-03-12 16:35:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779885","meta_rev_id":"d0213eb2b82004058a8403f68282998fc657fc67","_number":779885,"virtual_id_number":779885,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-12 16:35:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-03-12 14:15:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-03-12 15:04:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-03-12 15:04:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 02:36:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-12 14:15:13.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-03-12 15:04:11.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"56751053c9948dcb26bdb856dfa025603555cbb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-11 01:08:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8b754b8a9f0830a1a9f3ec7ce4c3059b9279a3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-11 01:09:18.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fc2c2300b4f7ae82d6406d9d6fa583d79e1e4221","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 02:36:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5910327b11004177b2d06e82b6caec72 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/389d277ff77540daa02799ece6fb2393 : SUCCESS in 4m 17s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0824bae143a34e9794f5a974fd09a30e : SUCCESS in 11m 45s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb093f6d16b547fa8204b4d1d10d8ded : SUCCESS in 10m 11s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3a158cce003487cb6ed6ffa8d7961df : SUCCESS in 8m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a00b888b3cea4f95818b8d967be6a98a : SUCCESS in 9m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f52c6c68d5f4820937961c57fb78525 : FAILURE in 41m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/afcdbfece38c46f3a303cff76ae75af5 : SUCCESS in 1h 21m 32s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d48403dd4a84558a67e2386a316ecd4 : SUCCESS in 1h 10m 04s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/c08b35d55d8840b79cefac641b29e12a : SUCCESS in 18m 45s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/b304ace56a2942fa8b57c81dc94b95da : SUCCESS in 17m 54s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a71b273364204e72963be802b74fbecf : SUCCESS in 16m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e89bdcdca346472c12770df3413a577421a71b7f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-03-12 14:15:13.000000000","message":"Patch Set 2: Code-Review-1\n\nremove TODO comment above","accounts_in_message":[],"_revision_number":2},{"id":"94f93d144be2bead4cbc14955e22518b4151e26d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-03-12 14:15:38.000000000","message":"Patch Set 2: Code-Review+2\n\nah, not required, changing default value","accounts_in_message":[],"_revision_number":2},{"id":"023fb41f711bd6f3642693c87f9a0027e2f92308","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-12 15:04:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d2f4d5f4fa4e7c94937ff8c2facc5843794bfe9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 15:06:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ac39882aa223469164cf0cbc017b329056e3dfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 16:35:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f1a858fbb7c43cd84694b0bbd033db5 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c2bd30407824221abf432274b4ab9be : SUCCESS in 3m 46s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b4dd876ccadd42548f6b2bc40f2931f0 : SUCCESS in 9m 15s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d446b9b6c0c43ce8c9d4f1dcf939a58 : SUCCESS in 7m 39s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/213efa2838c24d10be4a3c235337fa78 : FAILURE in 42m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/07c1f67d510149b5b2bef17cf7b26d5f : SUCCESS in 1h 24m 01s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/09ec1f6ef50443eabbf107690a8443c8 : SUCCESS in 16m 54s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/51d194af04d74d2f99cb54d90847958e : SUCCESS in 18m 20s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35dab5de4ab44a849c34590185a960f9 : SUCCESS in 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d0213eb2b82004058a8403f68282998fc657fc67","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 16:35:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"642fbc07a6857206a963064c9c2cd14030912226","revisions":{"b483fd2e936d812a6a0cb79eacfd747af8799df8":{"kind":"REWORK","_number":1,"created":"2021-03-11 01:08:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/779885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/85/779885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/85/779885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/85/779885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/85/779885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/85/779885/1"}}},"commit":{"parents":[{"commit":"e05af2b3f1dbb5ba74e5428ef3e74e4bfc096531","subject":"Use consistent names for service_credentials options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e05af2b3f1dbb5ba74e5428ef3e74e4bfc096531"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-11 01:06:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-11 01:06:48.000000000","tz":540},"subject":"Fix the error raised when password is not set","message":"Fix the error raised when password is not set\n\nCurrently, if the aodh::service_credentials::password parameter is not\nset then the following error is raised.\n\npick(): must receive at least one non empty value\u0027 if args[0].to_s.empty?\n\nThis change replaces the default so that we set the intended error\ninstead.\n\nChange-Id: Ia63c6e752a75d8acbcc3a2fc3b34c6ace65fc129\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b483fd2e936d812a6a0cb79eacfd747af8799df8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b483fd2e936d812a6a0cb79eacfd747af8799df8"}]},"branch":"refs/heads/master"},"642fbc07a6857206a963064c9c2cd14030912226":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-11 01:09:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/779885/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/85/779885/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/85/779885/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/85/779885/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/85/779885/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/85/779885/2"}}},"commit":{"parents":[{"commit":"e05af2b3f1dbb5ba74e5428ef3e74e4bfc096531","subject":"Use consistent names for service_credentials options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e05af2b3f1dbb5ba74e5428ef3e74e4bfc096531"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-11 01:06:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-11 01:09:05.000000000","tz":540},"subject":"Fix the error raised when password is not set","message":"Fix the error raised when password is not set\n\nCurrently, if the aodh::service_credentials::password parameter is not\nset then the following error is raised.\n\npick(): must receive at least one non empty value\n\nThis change replaces the default so that we set the intended error\ninstead.\n\nChange-Id: Ia63c6e752a75d8acbcc3a2fc3b34c6ace65fc129\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/642fbc07a6857206a963064c9c2cd14030912226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/642fbc07a6857206a963064c9c2cd14030912226"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
