)]}'
{"id":"openstack%2Fpuppet-aodh~784475","triplet_id":"openstack%2Fpuppet-aodh~master~Ibf56bdd22ac741e8fd4a101214fc9f477ed3c132","project":"openstack/puppet-aodh","branch":"master","topic":"policy-dirs","hashtags":[],"change_id":"Ibf56bdd22ac741e8fd4a101214fc9f477ed3c132","subject":"Allow to configure policy_dirs","status":"MERGED","created":"2021-04-01 20:18:21.000000000","updated":"2021-04-06 10:53:04.000000000","submitted":"2021-04-06 10:51:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784475-policy-dirs","meta_rev_id":"5521f59ca4f16f6f504dcef0a3706f158e255dff","_number":784475,"virtual_id_number":784475,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-06 10:51:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-04-06 09:38:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-04-05 21:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-04-06 09:38:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-01 22:18:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-05 21:25:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-04-06 09:38:07.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"aba40ec068c7c307e875596a9d7c89976857b839","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-01 20:18:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2098cdb747a0f13f28060c0dcdb89b35af5594e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-01 20:49:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e83849c154b39342a002c9688d2e452a1b42a52f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 22:18:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33d0ee85654e4c08a5957163195567d9 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c026eb4bf46497282e16e63d38e44d9 : SUCCESS in 3m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/83a8646fe96a41f4aa90d8c53700ebc4 : SUCCESS in 9m 52s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f78fb2677ffa48f69fa9ecb5608a9409 : SUCCESS in 7m 51s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09408a5d0f2f404ca9340003efb45d5b : SUCCESS in 7m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ce656c6bf4a408c9a21cb0c2ff2da01 : FAILURE in 29m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fe2382bd5eb46898672352c8f5ecb20 : SUCCESS in 1h 26m 06s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6a78995eaad439db0071baa2e1718be : SUCCESS in 17m 11s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/047ba5c78910420d8f3e0a962e93eb1c : SUCCESS in 17m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b9dd4e953e5489d9c9fba395bf2e62e : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"1e0011e508a6dda594458fec55ad46f16490cf6f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-05 21:25:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b25c786e31620b8420c79769b77c94e305dcf646","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-06 09:38:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3f8836fa609b9526ac733c9acbfcc605bed53adb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 09:39:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3455d2dc19a630d42b2a9db72d08dde93a6f0b10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 10:51:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7638ee2b100140ce895b99b0b1433598 : SUCCESS in 6m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08d9ad7d17f349068a9e6fc4d000ab64 : SUCCESS in 4m 31s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b67cd5e6e5f9431388af57fc520fb111 : SUCCESS in 9m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d270103a6f4496a824f95868e790738 : SUCCESS in 7m 06s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc2925b7d65547d6ae0d4d234fd70990 : SUCCESS in 1h 10m 40s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad1c5b3d37d14079ad77ddc6ea67ebba : SUCCESS in 17m 11s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c600f0c3f83f422c9d4209ca585f8fda : SUCCESS in 14m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c833c96f55634f8e9a76775805601041 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":2},{"id":"ff2c7bd8fe5f0dd16dcdceb1269d9573f80abfb1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 10:51:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5521f59ca4f16f6f504dcef0a3706f158e255dff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 10:53:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b1253a9e9e6403282ec9e779e5d4758 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4de74dd67f340e3637cd49e881e7aab4f8ef87fb","revisions":{"c9d3a289bc90ab8672af4beff500e1ff46392445":{"kind":"REWORK","_number":1,"created":"2021-04-01 20:18:21.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/75/784475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/75/784475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/75/784475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/75/784475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/75/784475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/75/784475/1"}}},"commit":{"parents":[{"commit":"53365e8b21793f35ac658c4a103fa24e33e2ab5b","subject":"Update sphinx to the latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/53365e8b21793f35ac658c4a103fa24e33e2ab5b"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-01 20:18:03.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-01 20:18:03.000000000","tz":120},"subject":"Allow to configure policy_dirs","message":"Allow to configure policy_dirs\n\nThis patch makes it possible to override the current service default,\nwhich is /etc/\u003cservice\u003e/policy.d .\n\nChange-Id: Ibf56bdd22ac741e8fd4a101214fc9f477ed3c132\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c9d3a289bc90ab8672af4beff500e1ff46392445"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c9d3a289bc90ab8672af4beff500e1ff46392445"}]},"branch":"refs/heads/master"},"4de74dd67f340e3637cd49e881e7aab4f8ef87fb":{"kind":"REWORK","_number":2,"created":"2021-04-01 20:49:52.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/75/784475/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/75/784475/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/75/784475/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/75/784475/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/75/784475/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/75/784475/2"}}},"commit":{"parents":[{"commit":"53365e8b21793f35ac658c4a103fa24e33e2ab5b","subject":"Update sphinx to the latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/53365e8b21793f35ac658c4a103fa24e33e2ab5b"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-01 20:18:03.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-01 20:49:44.000000000","tz":120},"subject":"Allow to configure policy_dirs","message":"Allow to configure policy_dirs\n\nThis patch makes it possible to override the current service default,\nwhich is /etc/\u003cservice\u003e/policy.d .\n\nChange-Id: Ibf56bdd22ac741e8fd4a101214fc9f477ed3c132\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/4de74dd67f340e3637cd49e881e7aab4f8ef87fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/4de74dd67f340e3637cd49e881e7aab4f8ef87fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
