)]}'
{"id":"openstack%2Fpuppet-aodh~787258","triplet_id":"openstack%2Fpuppet-aodh~master~Ibd5f33715c856aba5a6c34b74d3d49ceb09e5f28","project":"openstack/puppet-aodh","branch":"master","topic":"init-settings-no-openstack-settings","hashtags":[],"change_id":"Ibd5f33715c856aba5a6c34b74d3d49ceb09e5f28","subject":"Debian api_uwsgi_config: Switch to :ini_setting","status":"MERGED","created":"2021-04-20 16:14:40.000000000","updated":"2021-06-10 18:03:39.000000000","submitted":"2021-06-10 18:03:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"787258-init-settings-no-openstack-settings","meta_rev_id":"c698d79066fac94b90d8dad2b400d7e0676329c8","_number":787258,"virtual_id_number":787258,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-10 18:03:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-10 15:08:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-06-10 16:47:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-06-10 16:47:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 17:48:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-22 15:16:05.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-04-26 13:11:28.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"33faaec9d84bdb047fdc72fc527bbda3613a2c14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-20 16:14:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af53016fd22b710671226103d7a3283e4952274a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 17:48:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00f4f57aa86f40ad8299aef4ccf3dd2d : SUCCESS in 4m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b227117a17a4dd48f3c5e86e0d086a6 : SUCCESS in 5m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4d6e971a9d784521994f8499a6b0c05b : FAILURE in 9m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d808c35d1bb14f64948eaa30253b0ffa : FAILURE in 10m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a90f99d5fb6420ea3cdef56565377f9 : FAILURE in 8m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27f4a211e98e411c9232cce45ab68bee : FAILURE in 36m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8911d54c8f514939969544913393f128 : SUCCESS in 1h 27m 21s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/d3cba0fbb0ec4d27aa18e07b2190d53d : SUCCESS in 23m 48s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/727601162a524fe3a017ae3a2ab4ca8c : SUCCESS in 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22ac26ddb00b07c916d2d50ebe63212cce20424b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-20 20:27:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"804f0ac1a87440495e47be2086a59d0f3dedfee4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 21:35:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81b6de60682746aab07c110173e4fa57 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68d3b50c4e5f48eca1ddd02413247269 : SUCCESS in 4m 07s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fbac17046c2f4bd78d62a02220860e31 : FAILURE in 10m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb0f66f9dc7744ac9bf3462d2ccddff6 : FAILURE in 7m 59s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43a8ffc4f1964fb29a0e6ca555bcfe1e : FAILURE in 8m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/269f39259afd4a149048f64a2fc6ab65 : SUCCESS in 51m 55s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a0477f5996248c180cecf0e093fe184 : SUCCESS in 1h 05m 13s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab9d4a049b4d44cda4dd1211804eb02b : SUCCESS in 14m 52s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cfeda2c0a534d55ae1337906d1c4f06 : SUCCESS in 19m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"887a588933a41b1a2277bcb61c74045b78610c52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-21 16:12:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3ab157d3cbc779a686c02c2efab50e5fa44b0fde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-21 16:17:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ba665c6c8d3adee2b39157779ffd2d08631679ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 17:53:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75892c8826764668831b2071c0fa154c : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dc66090b5c043e78ac23743b9f812d5 : SUCCESS in 4m 27s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9637f8f3e1f94078a76b3b0b5c2eac12 : SUCCESS in 10m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f17bc5582c034a6e89588f3699f516ad : SUCCESS in 8m 19s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfd3e339f9da473cbb503a7b8c435eaa : SUCCESS in 8m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa77c87d4e944dbb9b7d8120f07f5c21 : FAILURE in 36m 06s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/855a6984202044c185360cb2b3ac7e4c : SUCCESS in 1h 29m 21s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc40b7e1fbe343299802f62208570cd7 : SUCCESS in 23m 49s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ca4966620a249989b8b14e6226110be : SUCCESS in 15m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a956da9c2cd0ac68e2558ceebb8af39f2afd5ac9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-22 15:16:05.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e26ed52ade253ed0ca47c6f4f7cc17e2000d7bd5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-26 13:11:28.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ebbfb9f2bcae98b736c4332dafbb2d8849dd6a5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 13:11:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bb1dfc8dc6e18ce1234fd882e3c5bdb7cee534d0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-26 13:18:38.000000000","message":"Patch Set 4: Code-Review-1 -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6c2f822a63ff4622e168b4fde2415b9b707a474e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-26 13:20:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4},{"id":"072ebb500be17176c385e4b5d3afb4da777d0fb9","tag":"autogenerated:gerrit:restore","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-26 13:20:38.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"a43cb222115f25779aa879da35a66499863f9775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 14:50:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bf66aaa4e444763835643d2b5482ca3 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d854407f8d9e4b3e820e5acd9ada71e4 : SUCCESS in 3m 28s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d6ade96712614fc39750350c79f76281 : SUCCESS in 9m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9affc9b5d0694010a725bcfddbd837a9 : SUCCESS in 8m 15s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e021bb1902d5459abd5af0f2819753d9 : SUCCESS in 6m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/084bc8b5dd0042e6b52934dac942ef72 : SUCCESS in 50m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f66507ab23a4027aedd6f3a1bcc6f99 : SUCCESS in 1h 23m 13s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/f97adc03f6d54794a78d97ddb3e038ed : SUCCESS in 18m 32s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f558f12f5f53408b8ae0356728cc17e0 : SUCCESS in 14m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f73d9c52316dab3ba09ba974cee064789583af1a","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-26 23:30:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a55928e631ac5706dd72da26695062f0bc511be4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-27 00:09:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dc09322712ad23694c553bbd6875c5a07d35932a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-10 15:06:46.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"d5b38d14c98cf274e55539bb35d8e812f6c2ac95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-10 15:07:22.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"e124b28bd5d5c89c6b1108cb7893d5b864054c94","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-10 15:08:36.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"502e6887024649fb1eb0eda15f955f44a3a63a53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 16:41:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a46db15587d414ba6458451f8abdd1e : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d9a6df9af9e4dea8617f5e045a01eb8 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9eb03743bd4f4673ba3140bd9c68cdef : SUCCESS in 4m 43s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e2fa54c585d4c0291b9df1ec134a3ca : SUCCESS in 12m 11s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e777b22bb434c79b7f03480058e2f4a : SUCCESS in 12m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a93bdb464cfc4f1e932f30c6a26f4be4 : SUCCESS in 8m 52s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed0ff06f892d447e9c15bd2b0fe51ea0 : SUCCESS in 11m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aff9dd5101f14daa8381842a01b8df40 : SUCCESS in 11m 03s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d106a137f725442da3d26a7b271fe624 : FAILURE in 37m 01s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/07ddf0af57b8463e8a52af26524fb601 : SUCCESS in 1h 27m 07s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0df9325e47fb4a76b5c5403dbe7cbed5 : SUCCESS in 1h 11m 53s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3ffb95d21694cab803aba94953ce877 : SUCCESS in 20m 04s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3618075c798416cbc1b7755dcb12276 : FAILURE in 18m 25s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/8bc5c053831544739278860010630d8b : SUCCESS in 19m 30s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd130cc6807d4840807d76dce257c79a : FAILURE in 13m 52s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a8a8cc68a4d4dd9b6f637bf47c16242 : FAILURE in 19m 43s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0e6c1fc83da5892388cfabba6c2e2e385233ce73","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-10 16:47:13.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"43d3d06687d10786c9d4bedab818da9916c8f925","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 16:47:57.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"928060c1146a7d82d8e8d3ae40648cb962c39f0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 18:03:39.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8f03f2acce048e888dd1bbc85e8fa20 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/affa931b337e48f5b80666b1b03ddb61 : SUCCESS in 4m 09s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4233bce23c454ab3afa75e664b62f306 : SUCCESS in 4m 09s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c2c5ec5011e42caa694adc542adf7c2 : SUCCESS in 12m 38s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/496ee9de999d4dcd91d4efe67dcd6179 : SUCCESS in 12m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66cca4d69e694b5aa29ab83a445d7c1d : SUCCESS in 9m 24s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2873acf28ffe4c23a531da04c6a4e9e0 : SUCCESS in 9m 00s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e75860b862154e51bd2bd59321a4527b : SUCCESS in 1h 12m 09s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c904dddc661148bd94ec196621a3e479 : SUCCESS in 23m 17s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e93517cf3d642248526b4039176ca85 : FAILURE in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c698d79066fac94b90d8dad2b400d7e0676329c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 18:03:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ef9d67832a3d363d3d6fd3b4d895713df4c34513","revisions":{"410898a089428da132e3dab5a89481fbedab61ea":{"kind":"REWORK","_number":1,"created":"2021-04-20 16:14:40.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/58/787258/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/58/787258/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/1"}}},"commit":{"parents":[{"commit":"e3fa374ac9433a5bdfe7b654f864e0f10a1aaa13","subject":"Merge \"Add support for aodh_api_uwsgi_config in Debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e3fa374ac9433a5bdfe7b654f864e0f10a1aaa13"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-20 16:13:29.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-20 16:13:29.000000000","tz":120},"subject":"Debian api_uwsgi_config: Switch to :ini_setting","message":"Debian api_uwsgi_config: Switch to :ini_setting\n\nOn each puppet run, I\u0027m seeing:\nvalue changed [\u002712\u0027] to \u002712\u0027 (corrective)\n\nTherefore, this patch switches from :openstackconfig to :ini_setting\nto make the 2nd run not changing anything.\n\nChange-Id: Ibd5f33715c856aba5a6c34b74d3d49ceb09e5f28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/410898a089428da132e3dab5a89481fbedab61ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/410898a089428da132e3dab5a89481fbedab61ea"}]},"branch":"refs/heads/master"},"fee316fc1dd46f3d49eaa016af9f0a8bde984e5b":{"kind":"REWORK","_number":2,"created":"2021-04-20 20:27:10.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/58/787258/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/58/787258/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/2"}}},"commit":{"parents":[{"commit":"e3fa374ac9433a5bdfe7b654f864e0f10a1aaa13","subject":"Merge \"Add support for aodh_api_uwsgi_config in Debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e3fa374ac9433a5bdfe7b654f864e0f10a1aaa13"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-20 16:13:29.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-20 20:26:50.000000000","tz":120},"subject":"Debian api_uwsgi_config: Switch to :ini_setting","message":"Debian api_uwsgi_config: Switch to :ini_setting\n\nOn each puppet run, I\u0027m seeing:\nvalue changed [\u002712\u0027] to \u002712\u0027 (corrective)\n\nTherefore, this patch switches from :openstackconfig to :ini_setting\nto make the 2nd run not changing anything.\n\nChange-Id: Ibd5f33715c856aba5a6c34b74d3d49ceb09e5f28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/fee316fc1dd46f3d49eaa016af9f0a8bde984e5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/fee316fc1dd46f3d49eaa016af9f0a8bde984e5b"}]},"branch":"refs/heads/master"},"1c0981aba785afac1ae84b740f7f009bc265873c":{"kind":"REWORK","_number":3,"created":"2021-04-21 16:12:47.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/58/787258/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/58/787258/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/3"}}},"commit":{"parents":[{"commit":"e3fa374ac9433a5bdfe7b654f864e0f10a1aaa13","subject":"Merge \"Add support for aodh_api_uwsgi_config in Debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e3fa374ac9433a5bdfe7b654f864e0f10a1aaa13"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-20 16:13:29.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-21 16:12:34.000000000","tz":120},"subject":"Debian api_uwsgi_config: Switch to :ini_setting","message":"Debian api_uwsgi_config: Switch to :ini_setting\n\nOn each puppet run, I\u0027m seeing:\nvalue changed [\u002712\u0027] to \u002712\u0027 (corrective)\n\nTherefore, this patch switches from :openstackconfig to :ini_setting\nto make the 2nd run not changing anything.\n\nChange-Id: Ibd5f33715c856aba5a6c34b74d3d49ceb09e5f28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/1c0981aba785afac1ae84b740f7f009bc265873c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/1c0981aba785afac1ae84b740f7f009bc265873c"}]},"branch":"refs/heads/master"},"874a158c35492792d5ff1054ebd71c71a83cf46c":{"kind":"REWORK","_number":4,"created":"2021-04-21 16:17:10.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/58/787258/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/58/787258/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/4"}}},"commit":{"parents":[{"commit":"e3fa374ac9433a5bdfe7b654f864e0f10a1aaa13","subject":"Merge \"Add support for aodh_api_uwsgi_config in Debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e3fa374ac9433a5bdfe7b654f864e0f10a1aaa13"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-20 16:13:29.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-21 16:16:59.000000000","tz":120},"subject":"Debian api_uwsgi_config: Switch to :ini_setting","message":"Debian api_uwsgi_config: Switch to :ini_setting\n\nOn each puppet run, I\u0027m seeing:\nvalue changed [\u002712\u0027] to \u002712\u0027 (corrective)\n\nTherefore, this patch switches from :openstackconfig to :ini_setting\nto make the 2nd run not changing anything.\n\nChange-Id: Ibd5f33715c856aba5a6c34b74d3d49ceb09e5f28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/874a158c35492792d5ff1054ebd71c71a83cf46c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/874a158c35492792d5ff1054ebd71c71a83cf46c"}]},"branch":"refs/heads/master"},"65fb8cb8ff33a1f1ba30372d1dbe21e9e179aa54":{"kind":"REWORK","_number":5,"created":"2021-06-10 15:06:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/787258/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/58/787258/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/5"}}},"commit":{"parents":[{"commit":"e3fa374ac9433a5bdfe7b654f864e0f10a1aaa13","subject":"Merge \"Add support for aodh_api_uwsgi_config in Debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e3fa374ac9433a5bdfe7b654f864e0f10a1aaa13"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-20 16:13:29.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-10 15:06:42.000000000","tz":0},"subject":"Debian api_uwsgi_config: Switch to :ini_setting","message":"Debian api_uwsgi_config: Switch to :ini_setting\n\nOn each puppet run, I\u0027m seeing:\nvalue changed [\u002712\u0027] to \u002712\u0027 (corrective)\n\nTherefore, this patch switches from :openstackconfig to :ini_setting\nto make the 2nd run not changing anything.\n\nChange-Id: Ibd5f33715c856aba5a6c34b74d3d49ceb09e5f28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/65fb8cb8ff33a1f1ba30372d1dbe21e9e179aa54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/65fb8cb8ff33a1f1ba30372d1dbe21e9e179aa54"}]},"branch":"refs/heads/master"},"ef9d67832a3d363d3d6fd3b4d895713df4c34513":{"kind":"REWORK","_number":6,"created":"2021-06-10 15:07:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/787258/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/58/787258/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/58/787258/6"}}},"commit":{"parents":[{"commit":"e3fa374ac9433a5bdfe7b654f864e0f10a1aaa13","subject":"Merge \"Add support for aodh_api_uwsgi_config in Debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e3fa374ac9433a5bdfe7b654f864e0f10a1aaa13"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-20 16:13:29.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-10 15:07:18.000000000","tz":0},"subject":"Debian api_uwsgi_config: Switch to :ini_setting","message":"Debian api_uwsgi_config: Switch to :ini_setting\n\nOn each puppet run, I\u0027m seeing:\nvalue changed [\u002712\u0027] to \u002712\u0027 (corrective)\n\nTherefore, this patch switches from :openstackconfig to :ini_setting\nto make the 2nd run not changing anything.\n\nChange-Id: Ibd5f33715c856aba5a6c34b74d3d49ceb09e5f28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ef9d67832a3d363d3d6fd3b4d895713df4c34513"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ef9d67832a3d363d3d6fd3b4d895713df4c34513"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
