)]}'
{"id":"openstack%2Fpuppet-aodh~790267","triplet_id":"openstack%2Fpuppet-aodh~stable%2Fussuri~I8c0e806112b947fccc6b0a6e47c3e0d3722a3e84","project":"openstack/puppet-aodh","branch":"stable/ussuri","topic":"expirer-stable/ussuri","hashtags":[],"change_id":"I8c0e806112b947fccc6b0a6e47c3e0d3722a3e84","subject":"Add support for aodh-expirer cron job","status":"MERGED","created":"2021-05-08 04:09:08.000000000","updated":"2021-05-10 19:17:56.000000000","submitted":"2021-05-10 19:16:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790267-expirer-stable/ussuri","cherry_pick_of_change":746241,"cherry_pick_of_patch_set":8,"meta_rev_id":"dfd20b7a27d3a91bfe85972ae174c587b0c6e827","_number":790267,"virtual_id_number":790267,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-10 19:16:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-05-10 09:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-05-10 17:53:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-05-10 17:53:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-08 05:10:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-10 09:30:42.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-05-10 17:53:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"3d2e4e1e37ecbd193461e93f239ecd9997b25dd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-08 04:09:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8f3764b55b060cb288402371c321bce4b22f2955","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 05:10:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a95d7a71041b4c53aba538191b3ed8d4 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5239a474a79469e84ca6034796c8aa6 : SUCCESS in 2m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/396c83bcb0a044798e2b261495392c32 : SUCCESS in 4m 03s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/8b62306c24604f998b68d130a4abfd42 : SUCCESS in 8m 32s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/81d87def72e7462689e8a1c18f77f390 : SUCCESS in 9m 49s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/0ec4755e955c404289aab6f3e3770d4c : SUCCESS in 8m 15s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/db711d521683438aa10b16cd6520a4b3 : SUCCESS in 8m 42s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c0d84c835cc4bfa837953ddc8af5faa : SUCCESS in 8m 54s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d27fdb15415a44afacd7dc3aa3db24d3 : SUCCESS in 9m 20s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1822586535b94f589970842bb0dbc801 : FAILURE in 29m 20s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/78cd7ae772494c32ba85e8f9567a008e : SUCCESS in 49m 59s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b01a03b8d3bb490ba8dd6df338bfd02b : SUCCESS in 52m 50s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ef209013db3f4787a02307712802621b : SUCCESS in 17m 26s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/22998468e4984354bc1c39f4453cd2a0 : FAILURE in 19m 00s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98d00d1b449746a7aec89e13d407528b : SUCCESS in 12m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/551d248cd0e441e69336fcd9a2b24d48 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a473921632a35cf1d59b80c3b3effdf2a4353ce0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-05-10 09:30:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f6ed34b5ebb4eeb7bd05d2fd9038b29ab2e5c45b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-10 17:53:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9f387c8c0e1d78d55f40ef46a56f172fe4e0cbad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 17:54:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b42c1af5fac5283064915c6841509957cb1321a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 19:16:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d75c6b01f6f54b80aba3e732b364cb17 : SUCCESS in 3m 38s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/159672141dc24deea0ca71f3949ee643 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7fe1901ce3b744d9915c10ae99a6f797 : SUCCESS in 3m 22s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ec41dc87494e4c98906cfa746e034901 : SUCCESS in 10m 54s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/21d9486c8b424442976c5e59843fd6a6 : SUCCESS in 11m 29s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5783a8c34cfc4bbcbce810ede8ffb1d0 : SUCCESS in 8m 46s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/440f092e0f1f4ff3a5761d8e8ec72b6d : SUCCESS in 1h 20m 22s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/51d33b4706bc4aff8a9ee6de910cdb3d : SUCCESS in 56m 29s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/83bc956c341d42f69cba2114d6f4f4f0 : SUCCESS in 18m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c0fc47da0264631a4ccb48168ac0a50 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7c6061429ffc13e507ddeae8d3fc43c341a6939e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 19:16:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dfd20b7a27d3a91bfe85972ae174c587b0c6e827","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 19:17:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f65a584c9b344c6cb1eac8a19029315f : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"127fba460858d633a015147c51e18b61ecf699c3","revisions":{"127fba460858d633a015147c51e18b61ecf699c3":{"kind":"REWORK","_number":1,"created":"2021-05-08 04:09:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/790267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/67/790267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/67/790267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/67/790267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/67/790267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/67/790267/1"}}},"commit":{"parents":[{"commit":"e71b518d68d2dbf8aa93f55b5fe7a9c399d70fcd","subject":"Add support for the interface parameter in authtoken middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e71b518d68d2dbf8aa93f55b5fe7a9c399d70fcd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-13 23:38:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-08 04:09:08.000000000","tz":0},"subject":"Add support for aodh-expirer cron job","message":"Add support for aodh-expirer cron job\n\nChange-Id: I8c0e806112b947fccc6b0a6e47c3e0d3722a3e84\n(cherry picked from commit 89da4b14fa7512a9cc0afd9748c5b1c539834091)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/127fba460858d633a015147c51e18b61ecf699c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/127fba460858d633a015147c51e18b61ecf699c3"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
