)]}'
{"id":"openstack%2Fpuppet-aodh~790268","triplet_id":"openstack%2Fpuppet-aodh~stable%2Fussuri~Ie3111d5a168e8d15f3e04d410a5a5c30b44324c2","project":"openstack/puppet-aodh","branch":"stable/ussuri","topic":"expirer-package-stable/ussuri","hashtags":[],"change_id":"Ie3111d5a168e8d15f3e04d410a5a5c30b44324c2","subject":"Manage aodh-expirer package by aodh::expirer class","status":"MERGED","created":"2021-05-08 04:09:52.000000000","updated":"2021-05-10 19:18:54.000000000","submitted":"2021-05-10 19:16:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790268-expirer-package-stable/ussuri","cherry_pick_of_change":755932,"cherry_pick_of_patch_set":2,"meta_rev_id":"83df1c58e18f5fc200a5d1ff30d1ab4e7df1e524","_number":790268,"virtual_id_number":790268,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-10 19:16:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-05-10 09:31:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-05-10 17:53:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-05-10 17:53:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-08 05:10:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-10 09:31:04.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-05-10 17:53:08.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"bb7e99a5ae6042b430e1e42cb2b54ecd75ba9fa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-08 04:09:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b72ed5b07dfd20877ab8d9bb532e8f7fbecc12c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 05:10:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/193c3d6985b54f3d839f7dab81133472 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52c118ce23f3497aaea8da641c7398c9 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63a192447618447d86fd5536f2b82eaa : SUCCESS in 3m 23s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/bdcb2f229bf04affa9b025bfd7623427 : SUCCESS in 8m 21s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/79f31de7e15c46e893879c228df94088 : SUCCESS in 9m 03s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/d7837cf7e7b04c76b445b8dac042320e : SUCCESS in 9m 21s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/1dcf422f9dbf40148211c45642591077 : SUCCESS in 9m 58s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2efa68409454418adbb83a98af2957e : SUCCESS in 7m 36s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/177cfc53c7f9462e8921a8aa9ca0bb9e : SUCCESS in 8m 48s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a624b0ff853840b8a783f1f4062b0b6f : FAILURE in 42m 09s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/96550b3fc52b4875841a02a9906e2b73 : SUCCESS in 55m 45s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d08127164acd49458b813144923cc974 : SUCCESS in 46m 05s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/8583f9611d064190a06561a994d64bd2 : SUCCESS in 20m 59s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/a564ba4f96dd40acbde41bc816635e33 : FAILURE in 20m 00s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2c6726754c0424a985e2f2b44d78ee3 : SUCCESS in 13m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a28e4847eb0344b4bbfdb354c184e2a3 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7d4d550955e92db0bb5935a47be7926d8e148804","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-05-10 09:31:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b8538c0e4537b2da4599ca1c6c67e844b15c9216","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-10 17:53:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1f64877a16fd457733d17c0ac0b061d009e0302f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 17:53:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a670360e3b3273580a6484f9df022aacdfd5999","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 19:16:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f956741357c4ed7bc394f3bdba70f7f : SUCCESS in 4m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/480cdf31962f427fa8fb32109f5029cd : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83fb59ab8cb846838727f2a73ef4b9c9 : SUCCESS in 4m 01s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4d7cd19fc014412da598b0b58664f4e3 : SUCCESS in 12m 40s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/06f33cf69a3149778c1fb364be5d6f5b : SUCCESS in 12m 45s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7821d46a0dff47b08d9a9e6dae5e2af4 : SUCCESS in 13m 50s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2440fb4be21a48058364cdb547108058 : SUCCESS in 1h 00m 35s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3d0d5647463949cd8d4a51b09baeaa4a : SUCCESS in 1h 05m 15s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/c1c761ba21304726917ea77a591d7bb7 : SUCCESS in 22m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6f727a1954a4f2e8becc738acc43968 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e381f78eb6aa79ee6c9ccbd391978cea2c2a72d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 19:16:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"83df1c58e18f5fc200a5d1ff30d1ab4e7df1e524","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 19:18:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c54494594e545899b742037867b4eb0 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe6ca076f352cb6469a26cfeac553f711dfc1a26","revisions":{"fe6ca076f352cb6469a26cfeac553f711dfc1a26":{"kind":"REWORK","_number":1,"created":"2021-05-08 04:09:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/790268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/68/790268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/68/790268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/68/790268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/68/790268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/68/790268/1"}}},"commit":{"parents":[{"commit":"127fba460858d633a015147c51e18b61ecf699c3","subject":"Add support for aodh-expirer cron job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/127fba460858d633a015147c51e18b61ecf699c3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-03 14:02:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-08 04:09:52.000000000","tz":0},"subject":"Manage aodh-expirer package by aodh::expirer class","message":"Manage aodh-expirer package by aodh::expirer class\n\nSince aodh-expirer command is included in a different package, we need\nto manage that package as well in the corresponding class.\n\nChange-Id: Ie3111d5a168e8d15f3e04d410a5a5c30b44324c2\n(cherry picked from commit 007d3a82f6bd72361066ea731a94a67c1b4d5f2b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/fe6ca076f352cb6469a26cfeac553f711dfc1a26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/fe6ca076f352cb6469a26cfeac553f711dfc1a26"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
