)]}'
{"id":"openstack%2Fpuppet-aodh~790991","triplet_id":"openstack%2Fpuppet-aodh~stable%2Ftrain~I8c0e806112b947fccc6b0a6e47c3e0d3722a3e84","project":"openstack/puppet-aodh","branch":"stable/train","topic":"expirer-package-stable/ussuri-stable/train","hashtags":[],"change_id":"I8c0e806112b947fccc6b0a6e47c3e0d3722a3e84","subject":"Add support for aodh-expirer cron job","status":"MERGED","created":"2021-05-13 00:51:25.000000000","updated":"2021-05-17 19:15:27.000000000","submitted":"2021-05-17 19:14:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790991-expirer-package-stable/ussuri-stable/train","cherry_pick_of_change":790267,"cherry_pick_of_patch_set":1,"meta_rev_id":"de76c776885fafec94bb733175f354f1a3b5aa0a","_number":790991,"virtual_id_number":790991,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-17 19:14:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-05-17 14:52:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-05-17 18:26:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-05-17 18:26:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-13 01:58:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-17 14:52:52.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-05-17 18:26:42.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"7acb84398afdbb293cbfae3a3ae6604ac35293a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-13 00:51:25.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"a89bd3d4365aa7b3ca45f2bb104a043c3d71dbca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 01:58:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eff18b584490403587abf815f6439814 : FAILURE in 4m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7ec14cab7e04bd18e2cd211e3f0e1c2 : SUCCESS in 4m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16d0c877d52143fd8e39c9ce3d7c1c0f : SUCCESS in 4m 20s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d4fe0712ee8f42e7bd7e4c8c9f7b10ba : SUCCESS in 11m 21s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/781ba885c51b434fb9240ac815bff46b : SUCCESS in 13m 52s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/f739dfa632534357bb820e30aa033255 : SUCCESS in 9m 54s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c41169a4d0d144b2946b6a3f009a7d9f : SUCCESS in 8m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df7ca275e0094f27844afe5e8bfab0ef : SUCCESS in 11m 33s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1a051faaf0f1445caf4a5ab4374a1126 : SUCCESS in 41m 40s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/d58bb4dab5b84c96a39f648c38624309 : SUCCESS in 1h 01m 05s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cce97085e3fa4628bc116d25b76d281c : SUCCESS in 45m 49s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/950e6d1ecd434a0d88fe69328328e46b : SUCCESS in 16m 53s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46b7a8a4699a48329b7c1c25993e905d : FAILURE in 7m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/695c0a93e44048a7b09b26ec4b96142e : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d61e32806507769de322e5516d83acd1eee09917","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-13 02:01:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bd71fb0d3dc107ccbb57ca9d63cd6782587dd31a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 02:48:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/579a22135d1e4d049ef2cfbfe53a2e53 : FAILURE in 4m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83ba4bab95d740759d86b1b39be88e54 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48268a692d144dd8ac561c2294b6e707 : SUCCESS in 5m 32s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/363de047359847d7a684b39a124b1e0d : FAILURE in 10m 03s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2c4f52c3e0714b479665ddf6543c6d4d : FAILURE in 9m 30s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/1878dc80d5034d8ab1b39da5d506dd3b : FAILURE in 9m 21s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9435e81c308e4968a9d97ae40b4e100f : FAILURE in 8m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f469b76bf1b4651956c4bb8cbae1872 : FAILURE in 11m 06s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/383c81d7e06948d797c8efa7ba5605c8 : SUCCESS in 36m 03s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/1df0bfa835434d6da14fed5b384f6d9b : SUCCESS in 44m 14s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3bd48a143f24462c8a2244400273dd9a : SUCCESS in 44m 53s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9efaedf9855c4d2da18b330e5a554421 : SUCCESS in 15m 51s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1d31739d07147adb123d659942c77fa : FAILURE in 9m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8ec57ca1b5c400f9359060ee789bcea : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":2},{"id":"024ae6b8efd1e475a8a59a84370d7b9b235a4e32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-13 02:52:02.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e9f2c12b1f4d2ce405f56c4d87d5f132f89ae248","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 03:55:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60c799a603d44065b750065bc7aa480f : SUCCESS in 5m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b4f0fe96c9b4df7861355a13af1461f : SUCCESS in 4m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82e7aece672347cca24c7699c4c45e35 : SUCCESS in 4m 59s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ee0cb38565ca48a884f34a48cc483e12 : SUCCESS in 19m 27s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4312e4cb596a49599d61e8b474fef711 : SUCCESS in 14m 40s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/a08ed10ac0984dc5854cc8c37b5f784d : SUCCESS in 14m 01s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e0cb664dfd84d1591692d96fbe749ce : SUCCESS in 8m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e78bc8c70cf4461b9a4ff14296b3b72f : SUCCESS in 10m 26s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/76366e8b386041628cea2758a733161f : SUCCESS in 43m 19s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/c2180034acd34ff39e61294a17bd6484 : SUCCESS in 58m 28s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/498a2f2e71854019b0cfaf2209663a87 : SUCCESS in 51m 38s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/58b670bdc1f04e9f92d09c489109ba13 : SUCCESS in 18m 53s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4de46c554114546903bf0b351249d26 : FAILURE in 7m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55510d6efd3942e883cd194d729867ef : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":3},{"id":"54c01df58889afa20e7135f0c92a8f5d90f1be20","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-17 14:52:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"509aab9897187d8b2a608b5d144ad7ca6e382073","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-05-17 18:26:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fcfe675b0367b548555a3c1fdb508054f2896f77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 18:27:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ec32e2b0263f8e338d57c6771e25b8f446967124","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 19:14:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16dc2c7567cc4725b26615e7e686e8b0 : SUCCESS in 5m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3bb54a6e8d644cb970fd79ed62a9e01 : SUCCESS in 6m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/442f1b3ebde745fa98173ca51875b658 : SUCCESS in 4m 59s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b2efc70339f14c21b6bbe0e68a4482f5 : SUCCESS in 12m 33s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57810fee719444ce9077776a1e2dda25 : SUCCESS in 9m 25s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4f0cda0cbf3d4e94b74d432b61f00eff : SUCCESS in 37m 08s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/c671b62e585a43daa7a369c6fcd915c2 : SUCCESS in 45m 34s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4dbef595fe61408fbb7fb75314884da7 : SUCCESS in 15m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a73bba4866d4ac289dd5127f33cd5e5 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":3},{"id":"11be150ad26f72ebaf46ce38b55e36b6d5bfbb20","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 19:14:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"de76c776885fafec94bb733175f354f1a3b5aa0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 19:15:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9807c73387d94140bd69773691539f56 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b9056246eccca47826de4568547420c4308cfa30","revisions":{"58f7483c57382bc7a078c57428f5ab08d45a5562":{"kind":"REWORK","_number":1,"created":"2021-05-13 00:51:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/790991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/91/790991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/790991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/790991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/790991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/91/790991/1"}}},"commit":{"parents":[{"commit":"ff3603ec358097c8fe75a357297a962570edb153","subject":"Expose vhost_custom_fragment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ff3603ec358097c8fe75a357297a962570edb153"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-13 23:38:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-13 00:51:25.000000000","tz":0},"subject":"Add support for aodh-expirer cron job","message":"Add support for aodh-expirer cron job\n\nChange-Id: I8c0e806112b947fccc6b0a6e47c3e0d3722a3e84\n(cherry picked from commit 89da4b14fa7512a9cc0afd9748c5b1c539834091)\n(cherry picked from commit 127fba460858d633a015147c51e18b61ecf699c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/58f7483c57382bc7a078c57428f5ab08d45a5562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/58f7483c57382bc7a078c57428f5ab08d45a5562"}]},"branch":"refs/heads/stable/train"},"e73c14ff21509b364c12fb520ff9bb06e8b8d845":{"kind":"REWORK","_number":2,"created":"2021-05-13 02:01:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/790991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/91/790991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/790991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/790991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/790991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/91/790991/2"}}},"commit":{"parents":[{"commit":"ff3603ec358097c8fe75a357297a962570edb153","subject":"Expose vhost_custom_fragment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ff3603ec358097c8fe75a357297a962570edb153"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-13 23:38:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-13 02:01:04.000000000","tz":540},"subject":"Add support for aodh-expirer cron job","message":"Add support for aodh-expirer cron job\n\nBackport note:\nFixed lint error caused by old lint packages in stable/train.\n\nChange-Id: I8c0e806112b947fccc6b0a6e47c3e0d3722a3e84\n(cherry picked from commit 89da4b14fa7512a9cc0afd9748c5b1c539834091)\n(cherry picked from commit 127fba460858d633a015147c51e18b61ecf699c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e73c14ff21509b364c12fb520ff9bb06e8b8d845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e73c14ff21509b364c12fb520ff9bb06e8b8d845"}]},"branch":"refs/heads/stable/train"},"b9056246eccca47826de4568547420c4308cfa30":{"kind":"REWORK","_number":3,"created":"2021-05-13 02:52:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/790991/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/91/790991/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/790991/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/790991/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/91/790991/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/91/790991/3"}}},"commit":{"parents":[{"commit":"ff3603ec358097c8fe75a357297a962570edb153","subject":"Expose vhost_custom_fragment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ff3603ec358097c8fe75a357297a962570edb153"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-13 23:38:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-13 02:51:58.000000000","tz":0},"subject":"Add support for aodh-expirer cron job","message":"Add support for aodh-expirer cron job\n\nBackport note:\nFixed lint error caused by old lint packages in stable/train.\n\nChange-Id: I8c0e806112b947fccc6b0a6e47c3e0d3722a3e84\n(cherry picked from commit 89da4b14fa7512a9cc0afd9748c5b1c539834091)\n(cherry picked from commit 127fba460858d633a015147c51e18b61ecf699c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b9056246eccca47826de4568547420c4308cfa30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b9056246eccca47826de4568547420c4308cfa30"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
