)]}'
{"id":"openstack%2Fpuppet-aodh~791228","triplet_id":"openstack%2Fpuppet-aodh~stable%2Ftrain~I063eb5439337d08e2380bb0ad233be266899e7c0","project":"openstack/puppet-aodh","branch":"stable/train","topic":"t-release","hashtags":[],"change_id":"I063eb5439337d08e2380bb0ad233be266899e7c0","subject":"Prepare the final stable/train release","status":"MERGED","created":"2021-05-13 13:17:15.000000000","updated":"2021-05-25 16:17:08.000000000","submitted":"2021-05-25 16:17:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791228-t-release","meta_rev_id":"7a1cfd1739b04f89039a32e1de156ae167e7dcc2","_number":791228,"virtual_id_number":791228,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-25 16:17:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-05-24 15:15:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-05-25 16:06:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-05-25 16:06:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-13 13:22:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-24 15:15:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-05-25 16:06:33.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"d3b1b44a576ae386e4c5c823c06385f3522050cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-13 13:17:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81059ed1006394cafdb135c71f8e09258c9e2aac","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-13 13:21:47.000000000","message":"Topic set to t-release","accounts_in_message":[],"_revision_number":1},{"id":"766241ddb32c4e912da7beaea35a8ec51fa1e7c6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-13 13:21:56.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"e74754e38375682fbda72ec48ec2adceb6a53e44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 13:22:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/c2c9c972b2774c57b5a6ba019ed8f3a6 : SUCCESS in 2m 51s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26a62ef267c84f229d61b743f0761c03 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"99bdc36361d8dc36c8f8012f047a55ee7656a104","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-23 13:20:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"6299a8691c3bc8746c7f72fd30afccd03af1ae53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-23 13:26:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/8af611fbe12b4bdbaf88603909e039ce : SUCCESS in 3m 02s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ceb549314a0f4b42a4504632d44faf67 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":2},{"id":"bb831e578d4533ef084ba731a8153e6ac098ce4a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-05-24 15:15:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6756857b514664f4985d8b3af680d0c35043cba7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-25 16:06:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"278b4a689bfcca2687cc6739a6a40b4f74ed6fc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 16:06:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2fa0f220ad41602a1e6061f3dcac833cce03de50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 16:17:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/27aff578b9824e4481266815e59208a7 : SUCCESS in 2m 46s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71561716803c4a64ae712377f8836756 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"7a1cfd1739b04f89039a32e1de156ae167e7dcc2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 16:17:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"09972d8b065df7a7390b430614dbc3c9e6928a65","revisions":{"ab9394eec760b7d5188173d2a0ba9bd718c14d1a":{"kind":"REWORK","_number":1,"created":"2021-05-13 13:17:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/791228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/28/791228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/791228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/791228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/791228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/28/791228/1"}}},"commit":{"parents":[{"commit":"ff3603ec358097c8fe75a357297a962570edb153","subject":"Expose vhost_custom_fragment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ff3603ec358097c8fe75a357297a962570edb153"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-13 13:17:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-13 13:17:06.000000000","tz":540},"subject":"Prepare the final stable/train release","message":"Prepare the final stable/train release\n\nThe stable/train is entering in extended manitenance. Because we have\nseveal changes merged since the last release, a new release should be\ncreated before that.\n\nDepends-on: https://review.opendev.org/791063\nDepends-on: https://review.opendev.org/791064\nDepends-on: https://review.opendev.org/791204\nChange-Id: I063eb5439337d08e2380bb0ad233be266899e7c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ab9394eec760b7d5188173d2a0ba9bd718c14d1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ab9394eec760b7d5188173d2a0ba9bd718c14d1a"}]},"branch":"refs/heads/stable/train"},"09972d8b065df7a7390b430614dbc3c9e6928a65":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-05-23 13:20:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/791228/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/28/791228/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/791228/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/791228/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/791228/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/28/791228/2"}}},"commit":{"parents":[{"commit":"ff9d23efe45e520479df3a422e9c7e2fec4c249a","subject":"Manage aodh-expirer package by aodh::expirer class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ff9d23efe45e520479df3a422e9c7e2fec4c249a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-13 13:17:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-23 13:20:53.000000000","tz":0},"subject":"Prepare the final stable/train release","message":"Prepare the final stable/train release\n\nThe stable/train is entering in extended manitenance. Because we have\nseveal changes merged since the last release, a new release should be\ncreated before that.\n\nDepends-on: https://review.opendev.org/791063\nDepends-on: https://review.opendev.org/791064\nDepends-on: https://review.opendev.org/791204\nChange-Id: I063eb5439337d08e2380bb0ad233be266899e7c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/09972d8b065df7a7390b430614dbc3c9e6928a65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/09972d8b065df7a7390b430614dbc3c9e6928a65"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
