)]}'
{"id":"openstack%2Fpuppet-aodh~806322","triplet_id":"openstack%2Fpuppet-aodh~master~Ibb82de966c2fae111da43b613c38a1cb0e2dfa8c","project":"openstack/puppet-aodh","branch":"master","topic":"auth-ut","hashtags":[],"change_id":"Ibb82de966c2fae111da43b613c38a1cb0e2dfa8c","subject":"Do not test detail of keystone::auth::service_identity","status":"MERGED","created":"2021-08-27 06:52:48.000000000","updated":"2021-09-02 15:17:06.000000000","submitted":"2021-09-02 15:17:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806322-auth-ut","meta_rev_id":"45d8a8da002f887d27e791523834309707b38ad1","_number":806322,"virtual_id_number":806322,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-02 15:17:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-09-02 06:12:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-09-02 14:54:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2021-09-02 14:54:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-27 07:18:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-02 06:12:59.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2021-09-02 14:54:38.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"147fde47889ad76ad51eeb87233b6d1a7f42fd21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-27 06:52:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4e34cf15a3c0cc80af6fdc0da4834bd544cfc48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-27 06:53:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7d8347aeaa2023655deca739b3db438338a1f257","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-27 06:55:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c0c5ddf0b1e5b387509425a5b2a56c2ed0089c81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 07:18:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1e9c66fae814250b47ccc17ea1eb70f : SUCCESS in 5m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e09fe9e44fe040c8bc48d583c71d7b08 : SUCCESS in 5m 24s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0638e51a53b403bba1b150f4e6130a7 : SUCCESS in 5m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d532edc6fb3c422fb3a413f97b4e8a48 : FAILURE in 15m 49s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/00b720ae4b724ff3aa9eeb592113623d : FAILURE in 16m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be920ac6758b4d0dbe311ac7ee76d115 : FAILURE in 12m 50s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d38473a4c8fc4e4d93ffc8336db91c55 : FAILURE in 11m 56s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/041a7f7101344d35902f6ac1129263d9 : FAILURE in 13m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"02c3e7dc183661d1873f2554579d9a012f982da8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-27 08:09:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5b31ef5d19635461a3770cb013c4494e71bba454","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 08:26:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/447db422b79843d9ab1ebbff1fac987b : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0ae5aaca09d4e3e8d7e440fd9e3a40c : SUCCESS in 4m 36s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85dacc6b8fd04468a6e05df82eb6acc6 : SUCCESS in 4m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/045ce876ba014fa894b60d959831ce71 : SUCCESS in 13m 41s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2cf063dfe3ee40038b4c6d5190a6048d : SUCCESS in 14m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7ee322e491245d6a312e7d7045689f1 : SUCCESS in 9m 25s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07727739ba5942e8ae81b30d22d1d1a9 : SUCCESS in 12m 01s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77c3bde6d0c04ef6a968165e9c5b4254 : SUCCESS in 9m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ac3e6ffed75f3e59e5425ecf3032fdbdb151e036","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-09-02 06:12:59.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"95d3cdbe3fd1303fd7c89bd732d43b6246db6c65","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-09-02 14:54:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"29594098f4f903775f6986c1c0c9e4792b2c0fa5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 14:55:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e579b25acd0f5665eaef0409bd9b1a768e697692","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 15:17:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee8da74d80564f9d8f38f424a4e9ec51 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e32a18fba0694f63bbfb459b3acb103a : SUCCESS in 5m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99dde899b4eb4d3cb6b806fc1cd7e3d1 : SUCCESS in 4m 06s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bae6a7d0e8a14aa6b47be890bc06499f : SUCCESS in 15m 10s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0893a7bcbdab433cbaaeab7eb4fea96b : SUCCESS in 14m 11s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3b123a131d941f7892aa3cff14c923d : SUCCESS in 9m 57s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b70bb90abb843b688fb928391337a41 : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":4},{"id":"45d8a8da002f887d27e791523834309707b38ad1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 15:17:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c6f85f7becaf69fec375c1296b07cf4c0cc1a8fd","revisions":{"8866245e4cd14b54a90b7de3a83a0a332461bb82":{"kind":"REWORK","_number":1,"created":"2021-08-27 06:52:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/806322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/22/806322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/806322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/806322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/806322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/22/806322/1"}}},"commit":{"parents":[{"commit":"8c2a3a43222b5fbd016f11c7c00c514b86b7ead0","subject":"Fix ignored parameters of aodh::auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8c2a3a43222b5fbd016f11c7c00c514b86b7ead0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-27 06:51:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-27 06:51:48.000000000","tz":540},"subject":"Do not test detail of keystone::auth::service_identity","message":"Do not test detail of keystone::auth::service_identity\n\n... but test interfaces of that resource type. This helps us avoid\ndirect breakage caused by any change in puppet-keystone.\n\nChange-Id: Ibb82de966c2fae111da43b613c38a1cb0e2dfa8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8866245e4cd14b54a90b7de3a83a0a332461bb82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8866245e4cd14b54a90b7de3a83a0a332461bb82"}]},"branch":"refs/heads/master"},"32f0636999fcfa216119d8e333caebe9adbcc842":{"kind":"REWORK","_number":2,"created":"2021-08-27 06:53:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/806322/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/22/806322/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/806322/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/806322/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/806322/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/22/806322/2"}}},"commit":{"parents":[{"commit":"8c2a3a43222b5fbd016f11c7c00c514b86b7ead0","subject":"Fix ignored parameters of aodh::auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8c2a3a43222b5fbd016f11c7c00c514b86b7ead0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-27 06:51:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-27 06:53:32.000000000","tz":540},"subject":"Do not test detail of keystone::auth::service_identity","message":"Do not test detail of keystone::auth::service_identity\n\n... but test interfaces of that resource type. This helps us avoid\ndirect breakage caused by any change in puppet-keystone.\n\nChange-Id: Ibb82de966c2fae111da43b613c38a1cb0e2dfa8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/32f0636999fcfa216119d8e333caebe9adbcc842"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/32f0636999fcfa216119d8e333caebe9adbcc842"}]},"branch":"refs/heads/master"},"bb60a80fe45a52c84697afc919426ee44973266f":{"kind":"REWORK","_number":3,"created":"2021-08-27 06:55:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/806322/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/22/806322/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/806322/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/806322/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/806322/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/22/806322/3"}}},"commit":{"parents":[{"commit":"8c2a3a43222b5fbd016f11c7c00c514b86b7ead0","subject":"Fix ignored parameters of aodh::auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8c2a3a43222b5fbd016f11c7c00c514b86b7ead0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-27 06:51:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-27 06:55:17.000000000","tz":540},"subject":"Do not test detail of keystone::auth::service_identity","message":"Do not test detail of keystone::auth::service_identity\n\n... but test interfaces of that resource type. This helps us avoid\ndirect breakage caused by any change in puppet-keystone.\n\nChange-Id: Ibb82de966c2fae111da43b613c38a1cb0e2dfa8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/bb60a80fe45a52c84697afc919426ee44973266f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/bb60a80fe45a52c84697afc919426ee44973266f"}]},"branch":"refs/heads/master"},"c6f85f7becaf69fec375c1296b07cf4c0cc1a8fd":{"kind":"REWORK","_number":4,"created":"2021-08-27 08:09:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/806322/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/22/806322/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/806322/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/806322/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/806322/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/22/806322/4"}}},"commit":{"parents":[{"commit":"8c2a3a43222b5fbd016f11c7c00c514b86b7ead0","subject":"Fix ignored parameters of aodh::auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8c2a3a43222b5fbd016f11c7c00c514b86b7ead0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-27 06:51:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-27 08:09:51.000000000","tz":540},"subject":"Do not test detail of keystone::auth::service_identity","message":"Do not test detail of keystone::auth::service_identity\n\n... but test interfaces of that resource type. This helps us avoid\ndirect breakage caused by any change in puppet-keystone.\n\nChange-Id: Ibb82de966c2fae111da43b613c38a1cb0e2dfa8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c6f85f7becaf69fec375c1296b07cf4c0cc1a8fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c6f85f7becaf69fec375c1296b07cf4c0cc1a8fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
