)]}'
{"id":"openstack%2Fpuppet-aodh~813316","triplet_id":"openstack%2Fpuppet-aodh~master~I843154a46ac61c8ea54a7b28a42ad154568a845e","project":"openstack/puppet-aodh","branch":"master","topic":"paste-ini-sep","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-10-20 03:15:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I843154a46ac61c8ea54a7b28a42ad154568a845e","subject":"Allow customizing separator for api-paste.ini","status":"MERGED","created":"2021-10-11 01:42:03.000000000","updated":"2021-10-20 03:17:14.000000000","submitted":"2021-10-20 03:15:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813316-paste-ini-sep","meta_rev_id":"e4c1ffb43abd4f42e73585c20a39b14e786725f7","_number":813316,"virtual_id_number":813316,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-20 03:15:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-10-20 01:15:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-10-18 16:10:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-10-20 01:15:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-11 03:03:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-18 16:10:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-10-20 01:15:31.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"50616a79d054be0d8687b0627d0c65aabc321a6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-11 01:42:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a2fa4ed19fc746ac8e3b2fe5da81efbe61dad7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 03:03:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae870c76e5ee454aa4597926ea335304 : SUCCESS in 4m 09s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02ca032c0bb34dd2baa51050d45da9fd : SUCCESS in 3m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d41f7220a0f44deb89cdbf320781d2b : SUCCESS in 4m 45s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c487a1041cd4be497a0406bc1af0453 : SUCCESS in 10m 18s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/000d9712662e438488e98acc74f511e3 : SUCCESS in 11m 11s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00d327014fa34a9890fce3a4a989d591 : SUCCESS in 9m 06s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa0a7ec0b69a4eca87aea819b4fc3cc5 : SUCCESS in 9m 22s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96015fa536d2474f801062446715f1a1 : SUCCESS in 9m 29s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcbed14fb22143a588bc8246e65442bf : FAILURE in 7m 38s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/60c85b5281bf482081a64f66f5d0a58c : SUCCESS in 1h 18m 46s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bc8f88d1aeb461ba39f45862418dbbb : SUCCESS in 15m 36s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e98035d8f39b4766a9a3c1b51e68486b : FAILURE in 29m 01s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4062159f02d14da3a39e3d3ccb07fc66 : SUCCESS in 15m 50s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/529eb67834a04284ba858eb536b08e1d : SUCCESS in 15m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a25e2ee21114d8aa7254489711314cb : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3b186ca7422964546f75bd13de95a598ad968d6b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-11 14:59:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d87a8e27ab7d53e7b0d91a609c3e6ed6c7673020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 16:21:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbba884cff0d42fd915cbbbc71d53594 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89796fd1ee5e494086fd137ee6b6d8dd : SUCCESS in 4m 16s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6226350f5af49729a8d06316f9b3ec4 : SUCCESS in 4m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c2cb9dc7bd2c4a038b0ee6a11ea345ad : SUCCESS in 11m 42s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1dd1f5a240ab43dfb12b06ef6379415d : SUCCESS in 11m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c8f5f2994d44915949d82e19f45edf1 : SUCCESS in 8m 44s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45d762daf71c40cbb023a51b952495f8 : SUCCESS in 9m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/119237bac2b04ebaa4f2b2c9ab7b5dd0 : SUCCESS in 9m 38s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5d5dc3675124261827140127cc777e9 : FAILURE in 8m 22s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/574d4c4e6a7e4651bb7de7648d25bfe6 : SUCCESS in 1h 17m 13s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14170d16c31d47d6b62ba40ada8999e2 : SUCCESS in 13m 21s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/db3e7b82a6a24a9881d8baaee7f2dc64 : FAILURE in 29m 57s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64e4582aff714a5aaaeef831e6325b38 : SUCCESS in 13m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ed0f9a5de7b4bda86e9243187fb0da0 : SUCCESS in 13m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f7245a0ea42429eba922d6c01ffbe11 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"87de46e43f2523ed5343800fe77de686b2e6274c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-12 12:09:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bdc15eb38ab9b9407eb06f0e6db09f85da8e393c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 13:32:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18b3c6c2fb8847f7a604a74356c4a42b : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cec3f5e9feed4535b82657da16ec4926 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3410be8f89f4ff4a4211406449f36f6 : SUCCESS in 4m 25s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2aa37df447dc46ac943ac4d333b5bcda : SUCCESS in 13m 55s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/338c70155b7d412488e995cc8bc68896 : SUCCESS in 14m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34cd705e0dcc4a019f7fd3f6ce9e3a52 : SUCCESS in 11m 04s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/211c2006e9774b99b76fdbb2dcfb372d : SUCCESS in 9m 31s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01631350d43d4da6bd4be9fbd31cd459 : SUCCESS in 9m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51ffbf0ae8d14e3b91553ade9be32094 : FAILURE in 9m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/903ae489df6647f7b508dcbf35c04de8 : FAILURE in 1h 20m 03s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0890fedc05f84424b0077e0e1238f1c1 : SUCCESS in 16m 33s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c47a3fdb82b54f01be33bda95ca03fcb : FAILURE in 29m 55s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f58f8aaa76154d6da06a86ea48b4e694 : SUCCESS in 14m 41s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/683872668cec468388659b33233594e9 : SUCCESS in 14m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bdb512ec6664fd7a6061c04406acb04 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"6667e054ad4646fffc8486f13f38b1badd532342","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-12 23:11:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"644823a7fd7c21cff4206498f72b8bd90238ac1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 00:10:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85a358b441cc40af965133d6b480d99a : SUCCESS in 4m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c64a51ee1eae4bfb917a9a37a7961680 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b48551725b143fe89a792680270ea8a : SUCCESS in 3m 48s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/04a4f1d3c5e94f8ab8d9b6bed6ffb01c : SUCCESS in 11m 18s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba10de761c364da685b23a988b27ea91 : SUCCESS in 11m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b21d33e5ca44ad7b54a63fc0ce7798c : SUCCESS in 9m 26s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de74341b01724e6c8b40a5d9bdd36463 : SUCCESS in 8m 58s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ee0ef55c73f47e9a62a81708a64bbc5 : SUCCESS in 11m 55s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2df848369d204a03b06066d1f85b8edc : FAILURE in 7m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b6c4af0774f34ae8b08eec0feb8b229d : SUCCESS in 56m 47s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cde3e8cc6b744a2ba9e6710cc1be000 : SUCCESS in 17m 16s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/5696ad6f183448798c9457d27369b93c : FAILURE in 29m 13s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b9ff45bb9f949a3ac3b37fc0d4691e2 : SUCCESS in 12m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be758c1dfba041918646d7ade95aaa13 : SUCCESS in 14m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f8e10e6afb44aa78bd4afdb6ecb321e : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"181290fddb4ce227ab3ad89598f37370da047a7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-15 03:03:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c8cf7dc3417ee295c172489ca2418d14fcfb4ab9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 04:29:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd0525be54e94e27b413944c52ff3f41 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4167770363f5458291c3d44675125cb6 : SUCCESS in 4m 07s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a48a91e57cb94e54899f2b3bfecda2c5 : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/82e57d04f5e04153a8dac2918e75c99c : SUCCESS in 12m 00s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4466ceaa18b74f998f37b61835c82266 : SUCCESS in 12m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f02499e90e049d9a0aa4973bf42ecd1 : SUCCESS in 9m 14s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3c48c0c10b34b4abd49518bc431e64e : SUCCESS in 9m 11s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/418ead5959614efe946e53ebfb787917 : SUCCESS in 8m 48s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bbe563917884775aab6953fcb8bd2b0 : FAILURE in 7m 31s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/991b4f43bad04dd3ba074ba6427e7fc0 : SUCCESS in 1h 20m 51s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82a92d9fd45b4dc99ad60c06d29176f6 : SUCCESS in 20m 39s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/aef2ed51dcd2439d94f12d3d37ff5b30 : FAILURE in 29m 24s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c54e047242748d28bfc59416cfb6778 : SUCCESS in 14m 00s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d551f987eb674c439404dbe89cf05075 : SUCCESS in 12m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7887fda6d10d43e8ae44276f7a9c8fe1 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":3},{"id":"73e4aca3f17f947a7ee5ffe862b27dadce1b63c8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-18 11:40:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"74ce52cf84f44c0e39e1c77e98e1144769819c31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 12:53:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c3143644dd44ebcb70d457fa4323c23 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24951bae07d642b2a322667361f81594 : SUCCESS in 4m 28s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5eb26530076498db451a3b1a8d20087 : SUCCESS in 4m 09s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca28a71d560a46b9aa91e610868c4417 : SUCCESS in 12m 51s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ee4f4d710b648db9fa8241521bfe59a : SUCCESS in 12m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bbdc84dcf294651b0129245b0d5fe51 : SUCCESS in 10m 59s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f31b4d29cacf44fb9b10d5b03ee0b7be : SUCCESS in 11m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cf66c897dcb449daf58610017069a0c : SUCCESS in 11m 08s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8847d2b6f49547acaa23a68e74d11e76 : FAILURE in 8m 47s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/72a42a8df6bc401eb2cc6b5f5af9f1e7 : SUCCESS in 1h 09m 47s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c76cd74c7c414711848305d5e4cc83bc : SUCCESS in 15m 25s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1467a55402834050862e68d36bb9bdfa : SUCCESS in 19m 53s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5e13acbb2eb4ac08a086cc19e2e48f1 : SUCCESS in 17m 25s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/526c7fb319d34462acc8acb29f0b2630 : SUCCESS in 15m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/082e6941aef743cfbbdd8e15bff3148b : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":3},{"id":"de764011a9f28569f337758328f1c8cf3dd2ac7d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-18 16:10:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c0ff847ca11b4ca01dae38a0341f456032f38fc5","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-10-20 01:15:31.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bfbd8b7a306e614db338834946c9d75ddc3b6273","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 01:16:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ab7bfa9f08056bac2dce66fc8902f68fcbf365e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 03:15:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ad40c2f5d16443db9e1d9a87be1aafa : SUCCESS in 4m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3db5afd548484d79b8b3a9bac1ff3c91 : SUCCESS in 3m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f499054ad3f74bddb5b3d3a5e30b304d : SUCCESS in 3m 45s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1143db084ae640329222f6f8a49dc2b7 : SUCCESS in 11m 23s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/35f3f138579a473cad08749600d579ab : SUCCESS in 11m 06s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a1cd2d6329d4051a5307d74e6eadadb : SUCCESS in 8m 47s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48d805ceaf4c40fd9c567d7d3b1d30e1 : SUCCESS in 9m 30s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fdc70006eb47449d969adca8d4609445 : SUCCESS in 1h 52m 24s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7a82cbf217a1422fbc1770978a358795 : SUCCESS in 18m 35s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb635422d843445f93eb5c66fe6690c0 : SUCCESS in 13m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e73b0af670734d5fbced625fcedde1fb : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":3},{"id":"d32736b4da3c7747d88d3554424f6ffb101e171c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 03:15:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e4c1ffb43abd4f42e73585c20a39b14e786725f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 03:17:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff588e3d4f0c411da3c60c8bfe141cc2 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e9a758a6f99a2969d120b3f2be44b121f74c1c58","revisions":{"d3f8e89931157c07fd59eec56f0f3441f892cd0f":{"kind":"REWORK","_number":1,"created":"2021-10-11 01:42:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/813316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/16/813316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/16/813316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/16/813316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/16/813316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/16/813316/1"}}},"commit":{"parents":[{"commit":"0c0ddd127c088ab759447c75f4020fe367d6c069","subject":"Update master for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0c0ddd127c088ab759447c75f4020fe367d6c069"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-11 01:31:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-11 01:31:39.000000000","tz":540},"subject":"Allow customizing separator for api-paste.ini","message":"Allow customizing separator for api-paste.ini\n\nThe api-paste.ini accepts not only \"\u003d\" but also \":\" and some services\nlike Barbican have been using \":\" for their default api-paste.ini\nfiles.\n\n[composite:main]\nuse \u003d egg:Paste#urlmap\n/: barbican_version\n/v1: barbican-api-keystone\n\nThis change allows users to use \":\" so that they can update the ini\nfiles with keeping it consistent with the default fules\n\nChange-Id: I843154a46ac61c8ea54a7b28a42ad154568a845e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d3f8e89931157c07fd59eec56f0f3441f892cd0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d3f8e89931157c07fd59eec56f0f3441f892cd0f"}]},"branch":"refs/heads/master"},"739bbfcf09005beff10114f40dd9dedc2b750e0a":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2021-10-12 12:09:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/813316/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/16/813316/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/16/813316/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/16/813316/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/16/813316/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/16/813316/2"}}},"commit":{"parents":[{"commit":"75892ddf3a7e19cdf9b1eb119f4c84a65a53207d","subject":"Bump version for the start of Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/75892ddf3a7e19cdf9b1eb119f4c84a65a53207d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-11 01:31:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-12 12:09:27.000000000","tz":540},"subject":"Allow customizing separator for api-paste.ini","message":"Allow customizing separator for api-paste.ini\n\nThe api-paste.ini accepts not only \"\u003d\" but also \":\" and some services\nlike Barbican have been using \":\" for their default api-paste.ini\nfiles.\n\n[composite:main]\nuse \u003d egg:Paste#urlmap\n/: barbican_version\n/v1: barbican-api-keystone\n\nThis change allows users to use \":\" so that they can update the ini\nfiles with keeping it consistent with the default fules\n\nDepends-on: https://review.opendev.org/813614\nChange-Id: I843154a46ac61c8ea54a7b28a42ad154568a845e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/739bbfcf09005beff10114f40dd9dedc2b750e0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/739bbfcf09005beff10114f40dd9dedc2b750e0a"}]},"branch":"refs/heads/master"},"e9a758a6f99a2969d120b3f2be44b121f74c1c58":{"kind":"REWORK","_number":3,"created":"2021-10-15 03:03:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/813316/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/16/813316/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/16/813316/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/16/813316/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/16/813316/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/16/813316/3"}}},"commit":{"parents":[{"commit":"75892ddf3a7e19cdf9b1eb119f4c84a65a53207d","subject":"Bump version for the start of Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/75892ddf3a7e19cdf9b1eb119f4c84a65a53207d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-11 01:31:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-15 03:03:17.000000000","tz":540},"subject":"Allow customizing separator for api-paste.ini","message":"Allow customizing separator for api-paste.ini\n\nThe api-paste.ini accepts not only \"\u003d\" but also \":\" and some services\nlike Barbican have been using \":\" for their default api-paste.ini\nfiles.\n\n[composite:main]\nuse \u003d egg:Paste#urlmap\n/: barbican_version\n/v1: barbican-api-keystone\n\nThis change allows users to use \":\" so that they can update the ini\nfiles with keeping it consistent with the default fules\n\nDepends-on: https://review.opendev.org/813614\nChange-Id: I843154a46ac61c8ea54a7b28a42ad154568a845e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e9a758a6f99a2969d120b3f2be44b121f74c1c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e9a758a6f99a2969d120b3f2be44b121f74c1c58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
