)]}'
{"id":"openstack%2Fpuppet-aodh~815296","triplet_id":"openstack%2Fpuppet-aodh~master~I1a5e49e72a5082cdeaff79b3739d4246e7f01879","project":"openstack/puppet-aodh","branch":"master","topic":"heartbeat_interval","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-17 02:01:32.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I1a5e49e72a5082cdeaff79b3739d4246e7f01879","subject":"Replace deprecated [coordination] heartbeat","status":"MERGED","created":"2021-10-25 13:14:57.000000000","updated":"2021-11-17 04:57:22.000000000","submitted":"2021-11-17 04:55:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815296-heartbeat_interval","meta_rev_id":"efbd5c21df56149db4e095e36d63060d0ae04a95","_number":815296,"virtual_id_number":815296,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-17 04:55:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-11-12 02:11:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-10-25 14:49:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-11-12 02:11:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-25 14:35:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-25 14:49:17.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-11-12 02:11:34.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"843e70056e185ba1922c3a9b6fd46dc1086e1f90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-25 13:14:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"984abae7931fcb4077fa5c5d21b740209f981256","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-25 13:15:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"beac2559f10678b8b272ba67f6e229615dd86a34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 14:35:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/281311836a7b47f09f8f46b82af967ca : SUCCESS in 5m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/654707b3c2a6473a98f4f43354c5ff6b : SUCCESS in 4m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b85245fbe2cb4a7c84d329b858865b51 : SUCCESS in 4m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/13c010aedfec48e2bfa614c267820b14 : SUCCESS in 12m 48s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc39d7d18883492ca0531e230ad654d2 : SUCCESS in 11m 34s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7dc17d2f1254daaaec346bb132e1463 : SUCCESS in 9m 28s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a40a874711f748aa87c32f60a512868b : SUCCESS in 9m 27s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/941867b89c424422b43bac14c912c886 : SUCCESS in 9m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a6f3165d9294abba3d1ee0e72ebf80d : FAILURE in 7m 55s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/76923c1d39aa418d8f97a3b292c287f1 : SUCCESS in 1h 13m 52s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d5856b19dee488298a53927fe6b44f6 : SUCCESS in 14m 39s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/74f5eb0d15da493aa4d8932c8097ba34 : SUCCESS in 19m 56s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3be75d04980146ffa3340d3966de904f : SUCCESS in 14m 11s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90fd886bd91542ce924cde12e94d5fcc : SUCCESS in 15m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc021def797e4b93888314f28a1cf8b3 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":2},{"id":"e243d2dd7653fc45001582244262b533d2ba4d55","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-25 14:49:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e48f57195e388d764b3eb41d90009c900967b640","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-11-12 02:11:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3adf6d79415304f0a364df72edd4043ecaeb7209","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-17 02:01:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2788f57c020164f91bbdc28b574054fc525b94e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 03:23:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87fb54201ba342909f150d7165ed8d07 : SUCCESS in 5m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16b55f7facea43d498973ed256c08a09 : SUCCESS in 5m 11s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8957b8c4d62046ccb379391d54d2441e : SUCCESS in 4m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/38ac2684c8354d9aba3b871b6a9e479c : SUCCESS in 16m 23s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/38ef55889baf4bacb9175eac22827aee : SUCCESS in 16m 11s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bbdc2d51a29475386c99bde739416fc : SUCCESS in 13m 30s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65e389678ac94a768c8c004f544d1d57 : SUCCESS in 12m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8438581dcaf4fe396e17fc6bf32f7ea : SUCCESS in 16m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1334f5b88f9a4ce188f50eac8f5cd861 : FAILURE in 1h 15m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/10fc3f4b40d14de6aa873f42f7899f9d : SUCCESS in 1h 10m 43s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a3c5f2ab8d0480fbc045028915bcf7f : SUCCESS in 17m 01s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d11085e4f8143d2a63db6593972da3f : SUCCESS in 20m 49s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cebc8e940be34d3a8b9d7480e0e49816 : SUCCESS in 17m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b176c7ef04ec433b9894836af449e848 : SUCCESS in 19m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bb25d325d8e4df49cee5042dda4b91e : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"cf108ac45cd8a13f2d3f089ded5ec71fc4929b9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 03:25:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b0c68b58359f7f7075486bdc839e6eb5bed5a6b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 04:55:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e007fea9d5834c8093a046b1af163335 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7891f176f1bb40ddadf1b6eefeada2a2 : SUCCESS in 4m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/752471af616745cc9e387d51740d10ef : SUCCESS in 4m 19s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/13557078518c4588b195efcbc37d7035 : SUCCESS in 13m 53s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/42cfb916f1674995abfbc25dce38645c : SUCCESS in 13m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c2c0dd4622f44e7a1d0cacff0bab33b : SUCCESS in 11m 23s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d882a9e4e1740698e5a315702d8d25d : SUCCESS in 11m 03s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4aeca5630af6406fb63670d18f05ba83 : SUCCESS in 1h 25m 45s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/429ac03d68d64dc4a791dd1135901aeb : SUCCESS in 20m 54s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1ede1005c56488594d1aa19df5eb354 : SUCCESS in 16m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4ab9241a31742e99e07244eaef8080d : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b239a14e36a7d10f1e47ced75bc8c3617d1393ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 04:55:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"efbd5c21df56149db4e095e36d63060d0ae04a95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 04:57:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab8b80370c4f4297aa028a25ac98fb54 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"49eb825204860d8465158a41c84f6af558a67b41","revisions":{"2542b86df1cdfe3133f42da3ea3b23572e35dcdd":{"kind":"REWORK","_number":1,"created":"2021-10-25 13:14:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/815296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/96/815296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/815296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/815296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/815296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/96/815296/1"}}},"commit":{"parents":[{"commit":"75892ddf3a7e19cdf9b1eb119f4c84a65a53207d","subject":"Bump version for the start of Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/75892ddf3a7e19cdf9b1eb119f4c84a65a53207d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-25 13:14:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-25 13:14:03.000000000","tz":540},"subject":"Replace deprecated [coordination] heartbeat","message":"Replace deprecated [coordination] heartbeat\n\n... by the new heartbeat_interval parameter.\n\nDepends-on: https://review.opendev.org/799600\nChange-Id: I1a5e49e72a5082cdeaff79b3739d4246e7f01879\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/2542b86df1cdfe3133f42da3ea3b23572e35dcdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/2542b86df1cdfe3133f42da3ea3b23572e35dcdd"}]},"branch":"refs/heads/master"},"49eb825204860d8465158a41c84f6af558a67b41":{"kind":"REWORK","_number":2,"created":"2021-10-25 13:15:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/815296/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/96/815296/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/815296/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/815296/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/815296/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/96/815296/2"}}},"commit":{"parents":[{"commit":"75892ddf3a7e19cdf9b1eb119f4c84a65a53207d","subject":"Bump version for the start of Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/75892ddf3a7e19cdf9b1eb119f4c84a65a53207d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-25 13:14:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-25 13:15:44.000000000","tz":540},"subject":"Replace deprecated [coordination] heartbeat","message":"Replace deprecated [coordination] heartbeat\n\n... by the new heartbeat_interval parameter.\n\nDepends-on: https://review.opendev.org/799600\nChange-Id: I1a5e49e72a5082cdeaff79b3739d4246e7f01879\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/49eb825204860d8465158a41c84f6af558a67b41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/49eb825204860d8465158a41c84f6af558a67b41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
