)]}'
{"id":"openstack%2Fpuppet-aodh~821022","triplet_id":"openstack%2Fpuppet-aodh~master~Ie0c61d21b5a64210c79765dc3ebd3c39d4d98c5d","project":"openstack/puppet-aodh","branch":"master","topic":"puppetlabs-apache","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-08 23:03:58.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"Ie0c61d21b5a64210c79765dc3ebd3c39d4d98c5d","subject":"Clean up direct dependencies on puppetlabs-apache","status":"MERGED","created":"2021-12-08 13:06:56.000000000","updated":"2021-12-09 01:43:23.000000000","submitted":"2021-12-09 01:43:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821022-puppetlabs-apache","meta_rev_id":"c815a8a672dc739977d32fb348f79dacb676b192","_number":821022,"virtual_id_number":821022,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-09 01:43:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-12-08 20:42:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-12-08 21:14:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-12-08 21:14:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-08 14:57:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-08 20:42:35.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-12-08 21:14:48.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"eef6d3ef1324a098d7fa8fe20ecd475c8bb18415","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-08 13:06:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c075d33da6c91282c4e6e23ec5188c5abcdda92b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-08 13:07:13.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"926d76eb6f65251bae448ffb9c15d39a4f6e5dc6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-08 13:08:13.000000000","message":"Topic set to puppetlabs-apache","accounts_in_message":[],"_revision_number":2},{"id":"0b16a7ef6b6981d9ff45c7ff3050aa0dec693048","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-08 13:33:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f9e7aaac267b871f7875250b3d5c8adec0627729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 14:57:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/2413ab6b460a4e4ab410664020b87d12 : SUCCESS in 3m 46s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a093b4efb613458bbdae564841f129e6 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/223c16180a1847cabab7b43b075ee462 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6914b662f23f40baade280c84d3bbcee : SUCCESS in 3m 34s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/06968b71df8d45ed860bee0bc94bb62c : SUCCESS in 13m 38s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/bdf4d9ad2a604de1becd6417170beeac : SUCCESS in 14m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91035e3930704ee6bf2ec478f47a71a2 : SUCCESS in 11m 48s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8f508004e484f1fafaa0b7441414678 : SUCCESS in 9m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3f3d331b0fc4c8f840e9a9e1620a143 : SUCCESS in 11m 38s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d9533d9afa241f69ad248a33bd11a72 : FAILURE in 33m 24s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/71dbd4f7753646dba37b45192b3a0afc : SUCCESS in 1h 18m 57s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/50b1e563fc4240a9a0633e7b4c80f6c4 : SUCCESS in 59m 09s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48fd55b36bdd4cbdbf42482a1d51e816 : SUCCESS in 16m 40s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3b9d0cbacbd4b18ac3811a4a5dd8648 : SUCCESS in 23m 40s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b3631bae2c245a7946867141a59a6ee : SUCCESS in 16m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77afb00cfe1b48a0913820ee5e2b619c : SUCCESS in 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"607d58775ff7711f1cb995206ef28a0756663f84","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-12-08 20:42:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4e9d96734fc427568e12ce28e6e71da1666e8656","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-08 21:14:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d2d26eb43c4fb05000dd02a299bb532ad7de581d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 21:15:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9c0f2d81411baa868859afa27759878f6e6d1d0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 22:57:56.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/600a228d8ec949b285afbbfe6b4f1752 : SUCCESS in 2m 27s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/174a5d2468dd4967b9df26e17abc38fa : SUCCESS in 4m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9aea77498fc64f00b8920e259817d2c2 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0e7ad3eab0d4b88af169206f72c3e55 : SUCCESS in 4m 46s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e896aed41d584dc1a107dc6ca982f799 : SUCCESS in 14m 50s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca24fb84202f4f2683a05f9eb60538e6 : SUCCESS in 12m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/181f093cf0094ae1b9ced20c0e053a65 : SUCCESS in 10m 50s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d416dfc7e66460ca5e9af569c1c4f03 : SUCCESS in 12m 26s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0962cabee80c42dfba2013a658b9da9f : FAILURE in 1h 40m 51s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/aaf92ae904404c359fb3d8da1d8143be : SUCCESS in 24m 39s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57fefe3b9b534367aef5ff344b03949c : SUCCESS in 16m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8ba43a489bee0ca8c730c2dd8bd1bcc602ef657c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-08 23:03:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"409843351daed5ec3a841a8a1eb8f26249759105","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 00:16:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/06b07de6fa3b4dca922ef51e55e7bda3 : SUCCESS in 2m 44s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e323816205148a399b2e0c96efc6fb3 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a6ac0832b7843bab0b24820c3bc3ec3 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5ea4daaf44f42ddb3ac2a08d40b06d2 : SUCCESS in 3m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e267d230e4a14140a62b48f4a94e6f5a : SUCCESS in 11m 41s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b735a9f05ff0464490c86e3a3d2f22e8 : SUCCESS in 11m 24s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c22f0ab9372406bbc8740fcdac22368 : SUCCESS in 8m 53s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a5e3257fad44842922c5fd22dc6420e : SUCCESS in 9m 11s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fc654c42ec04e088fafdb660efcb7cb : SUCCESS in 8m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d661caf5d18e4532a97a1492ebb81cbd : FAILURE in 32m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a6146a6514b468a9515749ae06550a4 : SUCCESS in 1h 10m 29s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5118c41fa88e4d7cbc3ec6290c42f4c3 : SUCCESS in 53m 30s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8afad6f5059496d90df186e731f36c9 : SUCCESS in 15m 10s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0377d28daed849439f76ef7814fcd7ba : SUCCESS in 20m 22s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/507f2b11e7924877802c88f193f15015 : SUCCESS in 15m 11s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39045665bdb64bd6a40a9c03423e59da : SUCCESS in 19m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3d946960da6654489edcf855960c1a9050bf8e2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 00:17:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"962e3a0fb91a3c4b52c1ebbc0ce6ebbee4621cca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 01:43:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/b90c35c5da8a4e1da1b89c54e4d05192 : SUCCESS in 2m 12s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ac37995ef4c443fbad691dcf07778f7 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97e51115a0fa486f990980a0c8b196e9 : SUCCESS in 3m 16s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ad431081aa34026b8484036aff51a4c : SUCCESS in 3m 04s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b14c1910a2a43b38cab6be4a2d20eff : SUCCESS in 11m 13s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4699d27964f848bba70c7a0d092f4ba7 : SUCCESS in 10m 44s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2196f3728b4341f6972149b834563534 : SUCCESS in 8m 13s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3e524c99e5e41068f53afe5819aa892 : SUCCESS in 8m 04s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/03ec539656ce45edb675b893ffdbb18e : SUCCESS in 1h 21m 48s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/42315b7ec3a54435bca8e1499e8dd3da : SUCCESS in 19m 31s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb9d553477e442a9b717c238971bbed3 : SUCCESS in 13m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c815a8a672dc739977d32fb348f79dacb676b192","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 01:43:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c0075580a5018ed4ea7a2ee4a092183e911f630b","revisions":{"91928393a1df3f181bda377d7f3cdc92cac3d176":{"kind":"REWORK","_number":1,"created":"2021-12-08 13:06:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/821022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/22/821022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/821022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/821022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/821022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/22/821022/1"}}},"commit":{"parents":[{"commit":"1b1401ec4fca25f533a6efee11c67701abbfb5eb","subject":"Merge \"Add missing dependency of aodh_api_paste_ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/1b1401ec4fca25f533a6efee11c67701abbfb5eb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-08 13:03:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-08 13:05:45.000000000","tz":540},"subject":"Clean up direct dependencies on puppetlabs-apache","message":"Clean up direct dependencies on puppetlabs-apache\n\nThis change removes direct reference to some classes in\npuppetlabs-apache. Details are explained beloew.\n\n- The api class doesn\u0027t need access to anything defined in\n  apache::params\n\n- The following classes are included by the openstacklib::wsgi::apache\n  resource type, and current inclusions are just redundant.\n\nChange-Id: Ie0c61d21b5a64210c79765dc3ebd3c39d4d98c5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/91928393a1df3f181bda377d7f3cdc92cac3d176"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/91928393a1df3f181bda377d7f3cdc92cac3d176"}]},"branch":"refs/heads/master"},"054108845ac2c8191fd413bb6e78af2b7123f3e2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-12-08 13:07:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/821022/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/22/821022/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/821022/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/821022/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/821022/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/22/821022/2"}}},"commit":{"parents":[{"commit":"1b1401ec4fca25f533a6efee11c67701abbfb5eb","subject":"Merge \"Add missing dependency of aodh_api_paste_ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/1b1401ec4fca25f533a6efee11c67701abbfb5eb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-08 13:03:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-08 13:06:58.000000000","tz":540},"subject":"Clean up direct dependencies on puppetlabs-apache","message":"Clean up direct dependencies on puppetlabs-apache\n\nThis change removes direct reference to some classes in\npuppetlabs-apache. Details are explained below.\n\n- The api class doesn\u0027t need access to anything defined in\n  apache::params\n\n- The following classes are included by the openstacklib::wsgi::apache\n  resource type, and current inclusions are just redundant.\n\nChange-Id: Ie0c61d21b5a64210c79765dc3ebd3c39d4d98c5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/054108845ac2c8191fd413bb6e78af2b7123f3e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/054108845ac2c8191fd413bb6e78af2b7123f3e2"}]},"branch":"refs/heads/master"},"c0075580a5018ed4ea7a2ee4a092183e911f630b":{"kind":"REWORK","_number":3,"created":"2021-12-08 13:33:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/821022/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/22/821022/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/821022/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/821022/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/821022/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/22/821022/3"}}},"commit":{"parents":[{"commit":"1b1401ec4fca25f533a6efee11c67701abbfb5eb","subject":"Merge \"Add missing dependency of aodh_api_paste_ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/1b1401ec4fca25f533a6efee11c67701abbfb5eb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-08 13:03:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-08 13:33:03.000000000","tz":540},"subject":"Clean up direct dependencies on puppetlabs-apache","message":"Clean up direct dependencies on puppetlabs-apache\n\nThis change removes direct reference to some classes in\npuppetlabs-apache. Details are explained below.\n\n- The api class doesn\u0027t need access to anything defined in\n  apache::params\n\n- The following classes are included by the openstacklib::wsgi::apache\n  resource type, and current inclusions are just redundant.\n\nChange-Id: Ie0c61d21b5a64210c79765dc3ebd3c39d4d98c5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c0075580a5018ed4ea7a2ee4a092183e911f630b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c0075580a5018ed4ea7a2ee4a092183e911f630b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
